)]}'
{"id":"openstack%2Foslo.middleware~700139","triplet_id":"openstack%2Foslo.middleware~master~I0f7296cfd8586b5b41c7a237561c4d931ee1e3d1","project":"openstack/oslo.middleware","branch":"master","topic":"base_python","hashtags":[],"change_id":"I0f7296cfd8586b5b41c7a237561c4d931ee1e3d1","subject":"tox: Trivial cleanup","status":"MERGED","created":"2019-12-20 10:05:19.000000000","updated":"2020-01-30 13:26:02.000000000","submitted":"2020-01-30 13:24:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700139-1580390673200-9ea6cd4b","meta_rev_id":"6f5eab049a642c13a1286bab693ceccc0220b199","_number":700139,"virtual_id_number":700139,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 13:24:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-01-14 11:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-01-30 11:25:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 11:25:46.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-30 13:24:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdebf184cace857875e1ba2dfdd6b981b4db940b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 10:05:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"795fb9a2d47815ad929cfcd02bd91ab2f6b7590b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 11:49:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ef189c838102420b99793b1d90c7b3ee : SUCCESS in 1h 29m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc371e178b20422186f014c891654e96 : SUCCESS in 1h 38m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0913350c60b4d2bb7276d781a7a3ef7 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6fc0de64c54bfd9765d48b7de9155b : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7430edaf8d7b4c318c1f0d9f62d7fd2e : FAILURE in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59f8355b347d49778c4ec107abd33faa : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/708e80d742bc4840a7552589f1611a9d : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39cc63faba844ee2875d6bbe90cdbad7 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e91d84b7a474c2e876e931c6b95e249 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f61f3e53a29b3400e11f51ba8a65c50cd7eaee1a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 12:53:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"afef64b3df8f2846ade516bfab85bf93520cbe8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 14:22:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3213d933177845c99de747227d7d15e8 : SUCCESS in 1h 28m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86dd9a838eba4be9a38eedd5074f5510 : SUCCESS in 1h 16m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7169875ac634f0586d21d2d683f632f : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa40cdf1c5948c985fff4d67b1a2179 : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d4cb3fc9af7e4b6c87ea375d75b12974 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd8a6e65a7b243e18b7709fc2676304f : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/627f85954b2945f4b078bed5894a2138 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a4c8700c0347d6874267ddfa04701d : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76e320d516854c8596980755749f30a0 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"70acc22be506c694e54145045f454a23ad0c03d2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:40:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8a7873c5561f1b737a442811ec81a4517114c6e2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-08 19:02:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"643b5ea12a967cb2549caef60cdbc6c0d0ae4de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:53:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e743893ee9a34a6f84173ce52b3731d6 : SUCCESS in 1h 50m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13565bfa7ea541689288cfb1e076ae78 : SUCCESS in 1h 32m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c817796cdbf9445884460f9de5e28f6f : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0eca23cecf41dfa30a53406f3dc6c1 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/563720979a3d4ee8b7dadd04d00cf26f : FAILURE in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2462f2688d28440abcc724a9b8ee2902 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff4f5d1b68404510a19df4b9b078a487 : FAILURE in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602e30e6b27741158f0ee4505663b2d8 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1072cd51a2c142ad918eb6d2907e4edb : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"d8c09dae82ad0814d0b0df8b465c017ee8879798","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-14 11:55:07.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ea769e4608e407151d5e0a4ddf984f4b0e949cac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 14:02:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f2e727a05593472ebc6962156dfe5d9b : SUCCESS in 2h 02m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b128bc5e8ce74e75ac7049d69dd55d6c : SUCCESS in 1h 53m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/087ad709e0844ed59d469cb2ce0897dc : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f46d1b33e5594b249d1590b178fa106b : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6124b85394d34620a207de733b87a4bd : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2117a7a803ec42e494ef0a638d182980 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/849d786749e84cc0b7229cb539b49a3a : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6ccd56900f420c8f4cead3b5741b4c : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aee2009074c84b0496ce31088c35ffa8 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"aa1abb31df69315fe2ebe56ae0ac9d8669cb601a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 11:25:46.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"47ed3c734d423aa580785c934894b26a57e9c885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 11:25:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fef4b00f76ce57154d85ba6ade1f384fc299cdd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:24:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0bc741cb7ef544c6904eab63a64d82be : SUCCESS in 1h 49m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0ecef2123984ec6a06fccaacf4f85c3 : SUCCESS in 1h 27m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5586cb10936147e29db3e5d60f60705c : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c2b4b69c59048de86cad34a9549cf44 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42cebfd0b91a4783a8f02a3e06d096bf : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2916459f247444258bf99b338af828e5 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/185e82ab2e03467b969a2de378f16783 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b52267fea28f40589d41d48d544cf76e : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d61bac1458c54784a1dede2399208714 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"92d21318c387e88c64e89071a96fcb300049d861","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:24:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"692c0174da4edd62986ecddd4e33b53fd7206a7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:26:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88245b0bb15949168ca96aa7a02acb1e : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a328202cfec4afdb8aef21d6fb43184 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c93f92b611c9186594df20a715c8ea88f6fb9f22","revisions":{"96ea18ffdc0681be2c8983853264b00fdaac6e1c":{"kind":"REWORK","_number":1,"created":"2019-12-20 10:05:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/39/700139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/39/700139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/39/700139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/39/700139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/39/700139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/39/700139/1"}}},"commit":{"parents":[{"commit":"7c47de583a1928dafd3a9e7796360a8b7e3fb6ce","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7c47de583a1928dafd3a9e7796360a8b7e3fb6ce"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:04:28.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:04:28.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\nUse the default \u0027install_command\u0027\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nChange-Id: I0f7296cfd8586b5b41c7a237561c4d931ee1e3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/96ea18ffdc0681be2c8983853264b00fdaac6e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/96ea18ffdc0681be2c8983853264b00fdaac6e1c"}]},"branch":"refs/heads/master"},"c93f92b611c9186594df20a715c8ea88f6fb9f22":{"kind":"REWORK","_number":2,"created":"2020-01-08 19:02:26.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/39/700139/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/39/700139/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/39/700139/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/39/700139/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/39/700139/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/39/700139/2"}}},"commit":{"parents":[{"commit":"7c47de583a1928dafd3a9e7796360a8b7e3fb6ce","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7c47de583a1928dafd3a9e7796360a8b7e3fb6ce"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:04:28.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-08 19:02:20.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\nUse the default \u0027install_command\u0027\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nChange-Id: I0f7296cfd8586b5b41c7a237561c4d931ee1e3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c93f92b611c9186594df20a715c8ea88f6fb9f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c93f92b611c9186594df20a715c8ea88f6fb9f22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
