)]}'
{"id":"openstack%2Foslo.middleware~742138","triplet_id":"openstack%2Foslo.middleware~master~Ifc5ee35d399c628604ec1644eb447936803c45f5","project":"openstack/oslo.middleware","branch":"master","topic":"oslo-pre-commit","hashtags":[],"change_id":"Ifc5ee35d399c628604ec1644eb447936803c45f5","subject":"Adding pre-commit","status":"MERGED","created":"2020-07-21 11:33:58.000000000","updated":"2020-10-05 17:02:35.000000000","submitted":"2020-10-05 16:54:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742138-1601916858866-f0a34f09","meta_rev_id":"a495dbb857668b14e28c51edc7c8f60a19935096","_number":742138,"virtual_id_number":742138,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-05 16:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2020-09-22 10:29:27.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":1,"date":"2020-09-28 15:19:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-05 15:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-10-05 15:44:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 10:29:27.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-09-28 15:19:58.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-05 15:44:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-10-05 16:54:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ac23a41446e600374a4657b951f8ff96a5478ca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-21 11:33:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7af051f2cbcaf2cfc7e0d994ebf666af92fee546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 12:10:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bfebb4ae92a4859809a87535054aa7d : FAILURE in 30m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f4a56dfeabe406db2ab37ab48b66a48 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be3ff9674084e00afbee04824648763 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17c7c375625f42b58e95752d18c4bef2 : FAILURE in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2655c88327bb4375ad58069741ff8d38 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb501f6b20a4a6792650d57005fb148 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ed7f5deb934b03b59998ce023ba4e9 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ee919babd573bc2c173e71c16e639e02962f10c8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-21 14:32:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a46d115f27ca8d3b3eb09c6d0401ea4b5b83c0a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 16:04:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07dc6a6101424be8a24e35022db4d65a : SUCCESS in 1h 28m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d909e63f391c446fbeea572ab6e1e4e8 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/729b6e02ba4847088feadd7c83007a69 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eed7bda1b56141e18f97c1704dcadc48 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4783506315284f44a317e8c5229b1ae7 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86cbcb6c426240be8b2fda007f9f17e7 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cedfeb95e71488ab9fa050a47eaae34 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e2d38bc9aef95e450066a5f0b05840b61eb217f5","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-22 08:39:20.000000000","message":"Patch Set 1: Code-Review-2\n\nWe should hold on in these patches for a while and observe oslo.cache for some time before spreading the change around all Oslo.","accounts_in_message":[],"_revision_number":1},{"id":"63ad52fff16d5b9c0e14460ea87321e04464f957","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-22 09:18:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89efa6feb8b0ad19525e2ebf74ded72aeeccae0c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-22 09:43:07.000000000","message":"Removed Code-Review-2 by Moisés Guimarães \u003cmoguimar@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"e963590293d94628a0f23e4aee3042d3896a3342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:28:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/63c30d1832e141f6a79acddce0415381 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4eef06f938b46d5ae9af01ba78e3b44 : SUCCESS in 1h 07m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c66a89f699f24ccea2f4a7c298170a31 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ed9ca806174e59a1be4bc3f05ee3ce : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83dd7155f0514833b339ace8b086f74f : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f858dfa4e4734c9f91ad0e8dac37ee0d : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e0fd101a4a948b49f53b2ec2df5c686 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a82d67976bcf4e138dd78d4d96608d71 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"20f38e158b035664ee4f3fa7fcbe115a02c49c58","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-22 10:29:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"942d5c831fca3a2e8c33a4d761651374bab6198c","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-28 15:19:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1f514acbca39f4cdc1498269af140a8b1bfb803b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-05 15:44:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6b77f8021bfdea0a390cf93c599b9fea89afb30c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 15:44:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf459ddb52655f0e83c531c66b7b84b550b6158f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:54:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0f9538d613d62c779f7716a78bb0eb67f21a61a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:54:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c3f34d97d30479f92ab33659230b93d : SUCCESS in 2m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae0e60383def47e7902ea292a6a86c01 : SUCCESS in 1h 06m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bf43b5fca8747c68f9993d6e9f3fd7d : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dce3a248ee74726a1ebb45c859e6664 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/952c3c437fa94111a11c7b7804781109 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a9ce749f2e64a98b3cb69511ddca047 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bedd056936d84f30ac1b9e9856c1119e : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27f609e6a58849a89a08dba83514c4e8 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"73c6a46c8c805e20a16db245830530259914e7a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 17:02:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70e1b257c7634aa9a80e81a17e666da4 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4c8563e9cac46639df41310e0d7aa66 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d023c60c7427a8e582491723c28350d44335f248","revisions":{"f6808ba3e2c8ea8a64107b20757ea42299e33493":{"kind":"REWORK","_number":1,"created":"2020-07-21 11:33:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/38/742138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/38/742138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/38/742138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/38/742138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/38/742138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/38/742138/1"}}},"commit":{"parents":[{"commit":"17ada34571427fbe3b44bcf35ef773c78ca70b5d","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/17ada34571427fbe3b44bcf35ef773c78ca70b5d"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:33:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:33:41.000000000","tz":120},"subject":"Add pre-commit","message":"Add pre-commit\n\npre-commit will be used to:\n- trailing whitespace [1].\n- Replaces or checks mixed line ending (mixed-line-ending) [1].\n- Forbid files which have a UTF-8 byte-order\n  marker (check-byte-order-marker) [1].\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs) [1].\n- Check for files that contain merge conflict\n  strings (check-merge-conflict) [1].\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements) [1].\n- Attempts to load all yaml files to verify syntax (check-yaml) [1].\n- replace tabs by space (4 by default)(remove-tabs) [2].\n- Run flake8 checks (flake8) (local)\n\n[2] https://github.com/pre-commit/pre-commit-hooks\n[3] https://github.com/Lucas-C/pre-commit-hooks\n\nChange-Id: Ifc5ee35d399c628604ec1644eb447936803c45f5\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/f6808ba3e2c8ea8a64107b20757ea42299e33493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/f6808ba3e2c8ea8a64107b20757ea42299e33493"}]},"branch":"refs/heads/master"},"d023c60c7427a8e582491723c28350d44335f248":{"kind":"REWORK","_number":2,"created":"2020-09-22 09:18:10.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/38/742138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/38/742138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/38/742138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/38/742138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/38/742138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/38/742138/2"}}},"commit":{"parents":[{"commit":"17ada34571427fbe3b44bcf35ef773c78ca70b5d","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/17ada34571427fbe3b44bcf35ef773c78ca70b5d"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:33:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-22 09:17:45.000000000","tz":120},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: Ifc5ee35d399c628604ec1644eb447936803c45f5\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d023c60c7427a8e582491723c28350d44335f248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d023c60c7427a8e582491723c28350d44335f248"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
