)]}'
{"id":"openstack%2Foslo.middleware~777593","triplet_id":"openstack%2Foslo.middleware~stable%2Frocky~I787a0276ec0a62bc9e2f068e4e4ee1219a306474","project":"openstack/oslo.middleware","branch":"stable/rocky","topic":"cap-bandit-1.6.2","hashtags":[],"change_id":"I787a0276ec0a62bc9e2f068e4e4ee1219a306474","subject":"[stable-only] Cap bandit to 1.6.2","status":"MERGED","created":"2021-02-25 13:54:29.000000000","updated":"2021-03-02 15:14:24.000000000","submitted":"2021-03-02 15:13:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777593-cap-bandit-1.6.2","meta_rev_id":"5865698c0f0cd40c28609a869e900a615651c852","_number":777593,"virtual_id_number":777593,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 15:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-03-02 10:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-03-02 10:12:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 15:10:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 10:12:46.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"471d662fd3b988f964af89b9677de6230e9b450b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-25 13:54:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f48f7ef23765b1f1aa4fb3c702d4115e73c9bd5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 15:10:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67a9192db2634dbeb394f0bf1f5d4b4d : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbf7f2c383724d18a28933a1d095f1ac : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/069d64c78525425da49fbc969b768f6f : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc40716cfa7940388b731c107bc5193d : SUCCESS in 5m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a32b03337ada4bd2b4563e3663b61669 : SUCCESS in 3m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a3a3bf0623a64dbab226b249a854984d : SUCCESS in 1h 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a20d39378a2a4c409c5e1b7eb158cbea : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c672a526fff42b68c16e8c70c134b83 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"69b9f877ef461ea8a3cae63ff50e68177419d3b2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-26 14:08:22.000000000","message":"Patch Set 1:\n\nThis patch fixes py27 stable periodic job failures: http://lists.openstack.org/pipermail/openstack-stable-maint/2021-February/thread.html#30762","accounts_in_message":[],"_revision_number":1},{"id":"cfc9ed8ac5f0efe8277174ab15c9b12e6f68eba7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-02 10:12:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial requirements change to align us with openstack/requirements. Fast approving.","accounts_in_message":[],"_revision_number":1},{"id":"c9f9664a48df6bbc2421059dac0b7a1d593825be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:13:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02fd108dd588b7ef33de26d15757eb9832928bb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 11:40:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d9bb8cc187439a88e545e65c3992e6 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35ee054909064129bd1dff5a73341e8a : FAILURE in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9f5684f413bd40ecb93e4680a9027909 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/330593733ab64fa89e0c1db22950ddd4 : SUCCESS in 5m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a87b23949f604d6483f0f8dd13d24943 : SUCCESS in 3m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e6b0d7a3996f4c8694825f7234ce5fc9 : SUCCESS in 1h 24m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63bddd65c42f4b21a8162905766ae0e0 : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8fa1fb4b4da4bc193ba64be75d74578 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"174137c58414426a150037940778972115963922","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-02 12:38:08.000000000","message":"Patch Set 1:\n\nrecheck - test_json_response","accounts_in_message":[],"_revision_number":1},{"id":"59cae55e2063a4b2456f3083c1d947144de951dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:53:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a674549e9e9349b2affae80b6447d8c3 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2ed504c591543b2befde9b73695b263 : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/24d723b1318a4a898cb4d7c204ee67ea : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efdc229a12ee4ccfb9af7ca114f51af1 : SUCCESS in 7m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dfae8c2058fa4873ad839bfba731fe5d : SUCCESS in 3m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e067eb42470143219d9b310132042acb : SUCCESS in 1h 14m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60453e912d6d4b4e8c9b661611a2ab05 : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43d72bb566c342d587444b464f850247 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"261afa2eb3d3c742eb3ab7eb8d9b52ae0e4ae997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:53:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aaa11434cd897ee4f4633dc8b440f3efe810cc32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 15:13:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9cc6c9ce48409b8b6a395b649f0fb0 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d63307963cfd4855b598aa3282ba4217 : SUCCESS in 4m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c113741861624f06a65cfcf9f5af1b29 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a885232df30445eca9b961e7a21e354f : SUCCESS in 4m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/80d5e084f10149f29a32cfad66d11eb0 : SUCCESS in 2m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9b991e0fac2645e6aa9e6e0aa0bb1773 : SUCCESS in 1h 18m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d6e690f8d604fae9b2f116b0a9d24b7 : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/156a68342f634d05b63cd63e9488dd11 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"40cdd2327eef759f9ecd0ac6c6cae4ac9f664016","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 15:13:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5865698c0f0cd40c28609a869e900a615651c852","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 15:14:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52eaaa9d982c4bc0aa5c0d922745856e : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9100ad4d86e0494da76bfcfd0d32c29f : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ba4498c475fb42b7f4074eaeebd587c8bc7eda6","revisions":{"9ba4498c475fb42b7f4074eaeebd587c8bc7eda6":{"kind":"REWORK","_number":1,"created":"2021-02-25 13:54:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/93/777593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/93/777593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/93/777593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/93/777593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/93/777593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/93/777593/1"}}},"commit":{"parents":[{"commit":"a0af15d9791ac71134aa2c25acba15831df35ae8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/a0af15d9791ac71134aa2c25acba15831df35ae8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-24 08:11:25.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-25 13:52:14.000000000","tz":60},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\nsphinx requirement needed to be updated to make requirements-check job\npass.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: I787a0276ec0a62bc9e2f068e4e4ee1219a306474\n(cherry picked from commit fba3b78fad2356f2715270a6f7d1ffbc27f63fd5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/9ba4498c475fb42b7f4074eaeebd587c8bc7eda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/9ba4498c475fb42b7f4074eaeebd587c8bc7eda6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
