)]}'
{"id":"openstack%2Foslo.middleware~779962","triplet_id":"openstack%2Foslo.middleware~master~I47d9016651f5cf14f316c47d1c145a79aa0bb08b","project":"openstack/oslo.middleware","branch":"master","topic":"fix_hacking_version_pre_commit","hashtags":[],"change_id":"I47d9016651f5cf14f316c47d1c145a79aa0bb08b","subject":"Fix the conflict status with hacking.","status":"ABANDONED","created":"2021-03-11 10:31:26.000000000","updated":"2021-03-22 09:46:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0c89e5095d50f0d60f7befe69e3774b5cf0f303d","_number":779962,"virtual_id_number":779962,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-11 12:24:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 12:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ac6359849db2c9a6b9f2904a67c9ad457ae1cf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-11 10:31:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a23ae025d5acd11b7d32d240b0d8a595d07910d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 12:24:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b394a86b6a24bdfab00304d54a2d44f : SUCCESS in 1h 45m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/821a175d4ed4439eb9271989bf98bb68 : FAILURE in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db4660bcd81f4e7c96b3d74d0c33b506 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b7d60703555443285b468be896f1694 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc1cb4e4423b41088b9b9d69092c59f9 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66dccc6fac4240cbb62b11db15678376 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d43e44883c407cb370031a9b470540 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0c89e5095d50f0d60f7befe69e3774b5cf0f303d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-22 09:46:47.000000000","message":"Abandoned\n\nUse the new approach[1] flake8 as local.\n\n[1] https://review.opendev.org/c/openstack/oslo.middleware/+/782114","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"516ec8e543ccf9282ab4d398eb73e220554a22e8","revisions":{"516ec8e543ccf9282ab4d398eb73e220554a22e8":{"kind":"REWORK","_number":1,"created":"2021-03-11 10:31:26.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/62/779962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/62/779962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/62/779962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/62/779962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/62/779962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/62/779962/1"}}},"commit":{"parents":[{"commit":"672919216bdf4209a2976e86f6e58850ffc9478f","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/672919216bdf4209a2976e86f6e58850ffc9478f"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-11 10:37:46.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-11 10:38:20.000000000","tz":60},"subject":"Fix the conflict status with hacking.","message":"Fix the conflict status with hacking.\n\nThe conflict is caused by:\n    The user requested flake8 3.8.3 (from /home/zuul/.cache/pre-commit/repozi7hytk7)\n    hacking 3.0.1 depends on flake8\u003c3.8.0 and \u003e\u003d3.6.0\n\nInstall the latest version of hacking to fix the conflict error with the\npre-commit hook.\n\nChange-Id: I47d9016651f5cf14f316c47d1c145a79aa0bb08b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/516ec8e543ccf9282ab4d398eb73e220554a22e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/516ec8e543ccf9282ab4d398eb73e220554a22e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
