)]}'
{"id":"openstack%2Foslo.middleware~785857","triplet_id":"openstack%2Foslo.middleware~master~Ic94f064f073755bee09495ced4f7627484bf8337","project":"openstack/oslo.middleware","branch":"master","topic":"stable/wallaby","hashtags":[],"change_id":"Ic94f064f073755bee09495ced4f7627484bf8337","subject":"Fix requirements issues","status":"MERGED","created":"2021-04-12 11:38:13.000000000","updated":"2021-07-12 13:52:05.000000000","submitted":"2021-04-15 14:54:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785857-stable/wallaby","meta_rev_id":"d1d59d777198e444d5a33a76a48d4e79668a33f9","_number":785857,"virtual_id_number":785857,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-15 14:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-04-15 13:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-04-15 13:47:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-12 13:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 13:47:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-07-12 11:23:51.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"33e07c27c844784590e73b22783bab88be426dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-12 11:38:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e7b8290d45703f0b646925a540fc66cbdef553a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 13:07:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f26c9daee6d47aebb7aa38879ddfe5f : SUCCESS in 3m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a229dda4037432d9188ff11d29245f3 : SUCCESS in 1h 27m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ce4bafb6c5648af92530fc1ac85bb64 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf31a32887b1450ab1184a58c946184e : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0539e0a95f54723884ddeb9f10b8fa6 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c0fd0d476e941b78ee1024d28c094f3 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1cd05fb83c9431c94b1f6b840091239 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20cd470fae2e4dc7b703a5885cd1c624 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"38989b6347db757704bb79aa048879603efefd06","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-15 13:47:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37aab8177be45a81df903ba8b756cb0df4c0dd82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 13:47:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"405f0905c14f3e3240df6c4eca77ec342568c01e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 14:54:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c0095cb6fa34c5ebb60ae71bd9d7543 : SUCCESS in 2m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b8be45651174b2385485f4d4af11c3f : SUCCESS in 1h 03m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dce477333e64c11b7837778ba6ae192 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7377033b018a4c2a8396786d9ca25633 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd4080715b3343d6a7b134f507746bee : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489dac0c823d467d9fc9450a0267a124 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1568bb7a97984281b26fbc1f6653bd94 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fd4cd4bb539e66beed4018b45f897b2cc9a78715","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 14:54:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44b2a6b7caa202fa83c953cdb84a606332590450","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 15:04:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4e9dc425f3444a085d89df515d2d916 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40e0919da59d443bb9ab78c7ccc794f4 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1},{"id":"8fbc704833d6586ed41504bcf397cfc9dce5e8a6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-12 11:23:51.000000000","message":"Patch Set 1:\n\nI don\u0027t really get how this was able to merge.\n\nThis is clearly a patch that was wrongly pushed towards master instead of stable/wallaby (as it shows in the topic). Moreover, the very same patch with different change-id was merged just before this patch ( I2f41e05df1cc978821bf7cb6d9a95bf543c7fba3 ).\nStrange :S","accounts_in_message":[],"_revision_number":1},{"id":"d1d59d777198e444d5a33a76a48d4e79668a33f9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-12 13:52:05.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I don\u0027t really get how this was able to merge.\n\u003e \n\u003e This is clearly a patch that was wrongly pushed towards master instead of stable/wallaby (as it shows in the topic). Moreover, the very same patch with different change-id was merged just before this patch ( I2f41e05df1cc978821bf7cb6d9a95bf543c7fba3 ).\n\u003e Strange :S\n\nI think it\u0027s because this was merged in and not rebased. The merge commit would have been essentially an empty commit. I missed that this was against master instead of stable/wallaby /o\\","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a0f7e456d3366a96b4b3e6c9f7d5acc8600f1a0","revisions":{"3a0f7e456d3366a96b4b3e6c9f7d5acc8600f1a0":{"kind":"REWORK","_number":1,"created":"2021-04-12 11:38:13.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/57/785857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/57/785857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/57/785857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/57/785857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/57/785857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/57/785857/1"}}},"commit":{"parents":[{"commit":"672919216bdf4209a2976e86f6e58850ffc9478f","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/672919216bdf4209a2976e86f6e58850ffc9478f"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-22 09:52:10.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-04-12 11:42:42.000000000","tz":120},"subject":"Fix requirements issues","message":"Fix requirements issues\n\nThis patch is the merge of 2 current fixes that need to be solved in the\nsame time. Indeed each issue lock our gates independently.\n\nDropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain needed\nto maintain that now exceeds the benefits provided by this mechanismes.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nMove flake8 as a pre-commit local target.\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach[1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements[2] will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nCo-authored-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n(cherry picked from commit 5c806d492b22f25b79e6598b2fe3847ba6963220)\nChange-Id: Ic94f064f073755bee09495ced4f7627484bf8337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/3a0f7e456d3366a96b4b3e6c9f7d5acc8600f1a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/3a0f7e456d3366a96b4b3e6c9f7d5acc8600f1a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
