)]}'
{"id":"openstack%2Foslo.middleware~805479","triplet_id":"openstack%2Foslo.middleware~master~I57408766c120d2c9784f3886457e91ee7d30de1b","project":"openstack/oslo.middleware","branch":"master","topic":"bug/1940738","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-10-26 13:23:11.000000000","reason":"Takashi Kajinami replied on the change"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-10-25 13:12:44.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"I57408766c120d2c9784f3886457e91ee7d30de1b","subject":"Add paste.filter_factory entrypoint for basic auth middleware","status":"MERGED","created":"2021-08-21 11:21:15.000000000","updated":"2021-10-26 16:22:28.000000000","submitted":"2021-10-26 16:20:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805479-bug/1940738","meta_rev_id":"1b0d6faf40e503e8e642fc6f90d216eeb9abf495","_number":805479,"virtual_id_number":805479,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-26 16:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-10-26 10:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-10-25 13:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-10-26 10:50:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-21 12:32:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-04 16:54:11.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-10-04 16:54:11.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"c35f7a6455b28b4494821a232c8c2a44af107f46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-21 11:21:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f15944c4a877472c910b90b3bbb7f98f9d38e6f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-21 11:23:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a68cf7ed4672ef1a230c533b00b4d8dfa826a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 12:32:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fa4707e45b694de6ab3819093f266b7c : SUCCESS in 4m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6cc62de900774a73a5bf9fc1ebbb738b : SUCCESS in 1h 09m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae056b57bf24ed18ec4bfe78ecbd98d : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93aa780b9333447aaa21cb0662ee731f : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e436b1fa89c41ab94b1cee08be13580 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c928a415c1574333b5fe50c783765f7c : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ea0d01774a478ab7a62935d0c952bc : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d5d17f5845680a281baab765a0d2fc7e81b7f970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-21 13:07:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"28e818b70255950f86a46519a215748d328eef43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 14:24:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b3cb5d5271e240b99be6bcc8f49ee973 : SUCCESS in 2m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1586156c5db64c598aab4ac4f246365a : SUCCESS in 1h 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29597263d4144715a62b64f5915c5468 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54272e860bdd4aa194e5e9ceeec3ab3d : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef116991d487434f925dc8990f5a756d : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c831589a5904b7d8a4c83ef34ca6ef4 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd2dc0c2499477d9cd2f89083f5a9bb : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"fab5c903858faf346bb634b12299028fb516e747","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-04 16:54:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3fa665640f3887b9b4cc71e2549bb3aadd68ab1d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-25 12:49:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b44be3cfe62afc97be2c05c08bede5dfbfb8c3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-25 13:06:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"83df3e122bed6bd7e77c26385e3b266fb2fd9659","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-25 13:07:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8965ecd82936065b0b59c0360ded3e0f376a6ba4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-25 13:12:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"55da7614f2a052ab308635438a0ec39d3c7a0859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 14:58:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2bdf2eb9b8834b89b1ef0e6db66bc909 : SUCCESS in 4m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ec139c5c31344818fd40dbeb081b2cb : SUCCESS in 1h 49m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d627c8067c64a3884f37cc382240c1f : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1332e33fd54d45ddbd3fad356a2c8ba3 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58e28c3ae4a44531acae99d2bbcf9b88 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1baf92c6fea4db18736423529ac1f54 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/424e3e415d854524b715537a67371f88 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":3},{"id":"2024a38b5cc5e257e75b907cdf200c175d72b907","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-26 10:50:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9bd686451ce8dcf7ed5a2300984e83b48e7b86ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 10:56:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"163120dd2424b54e85742d58badd2401f5bf3b6b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-26 11:07:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7f42287edac2a9b19c62e98a8ebe60e73060e67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 11:25:32.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3279849a3ae1460ca08197703e11f7ff : SUCCESS in 3m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e14af08c2cc4cacb4aa51e1ac164fb6 : FAILURE in 27m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaab20b847d74f4a92e55422ea74e21a : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbc2f80821eb4a0faa298914af92b129 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cad8bb6d32445f3bee7de2e76e7ad8c : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b946e389310e4bd89758f535adab2d73 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"f9ba1908f1b3a7d78979cb1d8099f39405809448","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-26 13:23:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"51fbce7bf63f94541da0970ec8e481c0292f1f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 14:57:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f7aaf3d0705049e69f5327ce395a9e2e : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f460bbd624214a6cb57e3cbdc10b043d : SUCCESS in 1h 31m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0dc952516a42dd97864937c03ddfd4 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44add3c02d01467697b8d81565fcdfc9 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9582522541274ef88494853841ab48c0 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/197c40787b324a79a59dd2803166c3f4 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44fbdd66bb3441479935f3da4a6fc6bb : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"365a748d6723fb8c4da6a4106f214d1c4a6e38cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 14:57:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c2aaea7f5fe8c6cb401d3b353f86c4ec67919509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:20:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/05b55752cfed4e36a19a9ab70b965973 : SUCCESS in 4m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c101a112c30a4ef99827ba42e53817da : SUCCESS in 1h 21m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b917b72a783e4d2b9221d5f5b1191e62 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bce286f8d1764d71a405aee5d3a84665 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a52bf87bc90444e19826fa54257b09e8 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb321821d864f07bbd411663fbac3db : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"7be37ce576a5566abbc9ca7dd868dd22ff77eba3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:20:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1b0d6faf40e503e8e642fc6f90d216eeb9abf495","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:22:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1acdad7020964dfc893168ca9ff11b36 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8e47bfc77c37610e0da577f13d7a3eb7894139a0","revisions":{"6bc6981bcfb0a5cc5f35e8bbcbe7a31dcb125984":{"kind":"REWORK","_number":1,"created":"2021-08-21 11:21:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/805479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/79/805479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/1"}}},"commit":{"parents":[{"commit":"da7987ca926e9bd82ff2989a920ea9740da24f95","subject":"Add new basic auth middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/da7987ca926e9bd82ff2989a920ea9740da24f95"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-21 11:19:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-21 11:20:57.000000000","tz":540},"subject":"Add paste.filter_factory entrypoint for basic auth middleware","message":"Add paste.filter_factory entrypoint for basic auth middleware\n\n... so that the middleware can be loaded by paste deploy mechanism.\n\nCloses-Bug: #1940738\nChange-Id: I57408766c120d2c9784f3886457e91ee7d30de1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/6bc6981bcfb0a5cc5f35e8bbcbe7a31dcb125984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/6bc6981bcfb0a5cc5f35e8bbcbe7a31dcb125984"}]},"branch":"refs/heads/master"},"0365cdbbc69a1341ffcff2aa95a1953364872f4e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-21 13:07:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/805479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/79/805479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/2"}}},"commit":{"parents":[{"commit":"da7987ca926e9bd82ff2989a920ea9740da24f95","subject":"Add new basic auth middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/da7987ca926e9bd82ff2989a920ea9740da24f95"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-21 11:19:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-21 13:07:29.000000000","tz":540},"subject":"Add paste.filter_factory entrypoint for basic auth middleware","message":"Add paste.filter_factory entrypoint for basic auth middleware\n\n... so that the middleware can be loaded using the python egg instead\nof the direct path.\n\nCloses-Bug: #1940738\nChange-Id: I57408766c120d2c9784f3886457e91ee7d30de1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/0365cdbbc69a1341ffcff2aa95a1953364872f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/0365cdbbc69a1341ffcff2aa95a1953364872f4e"}]},"branch":"refs/heads/master"},"8e47bfc77c37610e0da577f13d7a3eb7894139a0":{"kind":"REWORK","_number":3,"created":"2021-10-25 13:06:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/805479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/79/805479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/79/805479/3"}}},"commit":{"parents":[{"commit":"da7987ca926e9bd82ff2989a920ea9740da24f95","subject":"Add new basic auth middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/da7987ca926e9bd82ff2989a920ea9740da24f95"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-21 11:19:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-25 13:06:32.000000000","tz":540},"subject":"Add paste.filter_factory entrypoint for basic auth middleware","message":"Add paste.filter_factory entrypoint for basic auth middleware\n\n... so that the middleware can be loaded using the python egg instead\nof the direct path.\n\nCloses-Bug: #1940738\nChange-Id: I57408766c120d2c9784f3886457e91ee7d30de1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/8e47bfc77c37610e0da577f13d7a3eb7894139a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/8e47bfc77c37610e0da577f13d7a3eb7894139a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
