)]}'
{"id":"openstack%2Foslo.middleware~830040","triplet_id":"openstack%2Foslo.middleware~master~Iee839740756a53ee301e0a61d39b7caa9d78a31c","project":"openstack/oslo.middleware","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-20 02:14:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iee839740756a53ee301e0a61d39b7caa9d78a31c","subject":"Healthcheck: Reject requests with specific headers","status":"ABANDONED","created":"2022-02-20 15:07:07.000000000","updated":"2023-12-20 02:14:23.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92b6a5a29b2ae08ea573d81aeab67bc89149ecd2","_number":830040,"virtual_id_number":830040,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-20 16:10:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-20 16:10:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c94f6d8bf059c88cdc8640508fec2d34a0e4cc1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-20 15:07:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e1f53b9c5c8de944be98ba91aa24ad4bdb67226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-20 15:08:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ab25d8eb7a4f5128f2554f6bc06a397bf9d230ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 16:10:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8256bd1312f946459be62d4af00f833a : FAILURE in 1h 00m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f98195cc7e45cdab153815dc5cbded : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fafae7b30e95434ea267b117ea918483 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e850a0a8426447c912abe55e5f28ddb : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee944f167ca142468cb9b1da5639acf5 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17bb82d81ffd473ba3d9d1bf5b1191e5 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"3c11f26419b68d83ffa9cde6c38efbe56d02bb36","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-20 17:08:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23929216934e32a0627ea4626417dcab2cf95750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 18:12:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1517dc1ed41d42e487c7792fa788e23a : FAILURE in 1h 00m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70bdbdbc2b644dab8a0aa4cc5661c608 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44052adb8d9042799965149be4ab09db : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87a9cda35e72435eaef611f73d556346 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3dd1fd05924dbaa356a099366495a7 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22f48a0ed1904b71ad5611070ccfe216 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"92b6a5a29b2ae08ea573d81aeab67bc89149ecd2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-20 02:14:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86de48ba77fff7fa68d3136bb16b21e26770d74f","revisions":{"63e6a2292da709a4f9210adf0f4fc30a28399831":{"kind":"REWORK","_number":1,"created":"2022-02-20 15:07:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/830040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/40/830040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/40/830040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/40/830040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/40/830040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/40/830040/1"}}},"commit":{"parents":[{"commit":"2f72b30cbabef86be3ca91c45346f3d9ec92a75a","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2f72b30cbabef86be3ca91c45346f3d9ec92a75a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-20 15:01:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-20 15:01:23.000000000","tz":540},"subject":"Healthcheck: Reject requests with specific headers","message":"Healthcheck: Reject requests with specific headers\n\nThis change introduces the new prohibited_headers option, which defines\nlist of request headers. When a healthcheck request has any of these\nheaders, the healthcheck middleware rejects that request and returns\n404 Not Found.\n\nThis allows filtering external requests using the request headers\nappended by frontend loadbalancers(e.g. X-Forwarded-For), to prohibit\naccess to the healthcheck endpoint by external clients.\n\nChange-Id: Iee839740756a53ee301e0a61d39b7caa9d78a31c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/63e6a2292da709a4f9210adf0f4fc30a28399831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/63e6a2292da709a4f9210adf0f4fc30a28399831"}]},"branch":"refs/heads/master"},"86de48ba77fff7fa68d3136bb16b21e26770d74f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-20 15:08:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/830040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/40/830040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/40/830040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/40/830040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/40/830040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/40/830040/2"}}},"commit":{"parents":[{"commit":"2f72b30cbabef86be3ca91c45346f3d9ec92a75a","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2f72b30cbabef86be3ca91c45346f3d9ec92a75a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-20 15:01:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-20 15:07:13.000000000","tz":540},"subject":"Healthcheck: Reject requests with specific headers","message":"Healthcheck: Reject requests with specific headers\n\nThis change introduces the new prohibited_headers option. This option\ndefines list of request headers, and when a healthcheck request has any\nof these headers, the healthcheck middleware rejects that request and\nreturns 404 Not Found.\n\nThis allows filtering requests via front-end load balancers using\nthe request headers appended automatically(e.g. X-Forwarded-For), to\nprohibit access to the healthcheck endpoint by external clients.\n\nChange-Id: Iee839740756a53ee301e0a61d39b7caa9d78a31c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/86de48ba77fff7fa68d3136bb16b21e26770d74f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/86de48ba77fff7fa68d3136bb16b21e26770d74f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
