)]}'
{"id":"openstack%2Foslo.middleware~855946","triplet_id":"openstack%2Foslo.middleware~stable%2Fussuri~I467f8105ebf13557ee1fb539d3d36e4d1e968118","project":"openstack/oslo.middleware","branch":"stable/ussuri","topic":"cap-virtualenv-py37","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-19 18:22:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I467f8105ebf13557ee1fb539d3d36e4d1e968118","subject":"[stable-only] Cap virtualenv for py37","status":"MERGED","created":"2022-09-05 15:09:13.000000000","updated":"2022-09-19 18:26:47.000000000","submitted":"2022-09-19 18:22:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855946-cap-virtualenv-py37","meta_rev_id":"0def1ebbca778afc394bba8baa0aea2b2ea909cb","_number":855946,"virtual_id_number":855946,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-19 18:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-09-14 18:08:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-09-19 16:50:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-05 16:35:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-14 16:18:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-09-14 18:08:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"deb950cf00e951438e79225e1dd0fbb9bf7b6d17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-05 15:09:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7855fdd6fe21a1c3466a31ed83588ca22420752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 16:35:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee6f4932b64d49baa71c77eb9782f7d6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fc77128e0fc4f46b8fd75b66740032b : SUCCESS in 1h 25m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcbe22c293594291a23080bb5bc5d0ac : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ef7eec98a2420eb22c12e52150b59a : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb315035b2f74a408ef22af19d29d1a7 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20850ddb54604b739b9c31accf8da44c : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c76f8c10d74d0891d5b5fa99085915 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/474ef08e1de440019d9521ebf0ac8b89 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5dce0aacbdc296d7cf8e049952c8cd6e09e2fe7a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-14 16:18:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cde7a8fc5b6509291336fb7bb779395bd739d063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 17:33:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07997b1348ea4f5092c1539a8170f7e6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2181a6ac74ef41cfa9fefe83b81be73e : SUCCESS in 1h 13m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dc09bd81b004e2794b9e5969fd83f69 : FAILURE in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa3060813edf4c888e002d5448c9f081 : SUCCESS in 2m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aadf9c0f74e49ebbb8c56ffbe5602c5 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78e68eac4cd243ff8e27de4b20df1c69 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c71abeaa45412d96a94075e4ea7507 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f39ea0b74044a06970b36a625a5f1f0 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"31b1504b6090d0e8886fb9d797faf9907d488aef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-14 18:02:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2d8796df86f7e441a605cb236bd30ec5af663796","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-14 18:08:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d274a0130473cbc595122ba5bf450d380ba527f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 19:20:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e808910e02c14a96b7a444c3ec73fc9b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba9e0104df744ccdb7e3734de8e51080 : SUCCESS in 1h 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ca4bfce2414e65884195b663caf385 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f2245bfd7df4ad3be866dba570e4535 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfdaaf7fdce9453d8c948349de2fa6a9 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1484cdcc8064077a98ab7e6bf395e83 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50aba3ec769944c7aa7731dd627676ce : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"930770d1d499c730bcb9ee3c773bf5783c48b63e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-19 16:50:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c5ccb69b4c7d85432b66ed64b3d0696e7cedf32b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 16:51:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"55d4a5b80bfb257590a7f2314ec17c06707eab07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 18:22:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eff08ca628143668fa85b5005bf16e6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64f0c9cc995c4b949a862f4701786802 : SUCCESS in 1h 25m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed0504c3e6cf4182b0afbc22e53d3e71 : SUCCESS in 2m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2343658d653b4665837145663cf6f169 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5664dcc3c61466da1a75fb5a77af60c : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7550902ecdab488cb73d9e98f0b6a0c5 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4c3fb2df6f7476885963af6e9500def : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"02f7edad7127e0cb0cb42ccf37239fba0cd55cfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 18:22:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0def1ebbca778afc394bba8baa0aea2b2ea909cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 18:26:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e01f3f040ce4e9c8624a899d9dcfbd4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0eb6fc6d2b45e7a91667e399c6dbd0 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00c5a26d48944262a29947ee025ed58b : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"894b912d828c7746bf8c4ce4f2416500ad0f9b1a","revisions":{"39b9dcc20fc2bf785ac4cc393f12ea849bc2670f":{"kind":"REWORK","_number":1,"created":"2022-09-05 15:09:13.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/46/855946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/46/855946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/46/855946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/46/855946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/46/855946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/46/855946/1"}}},"commit":{"parents":[{"commit":"2ca70cec88f789fbcbcc9075c2253e30d14f3288","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2ca70cec88f789fbcbcc9075c2253e30d14f3288"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-05 15:08:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-05 15:08:25.000000000","tz":120},"subject":"[stable-only] Cap virtualenv for py37","message":"[stable-only] Cap virtualenv for py37\n\npy37 test started to fail with latest virtualenv (20.16.4) as it\nbundles setuptools 65.3.0, which causes installation problems under\npy37. (Note: with py38 the problem does not occur).\n\nThis patch sets virtualenv\u003c20.16.4 to unblock the gate as\nvirtualenv 20.16.3 bundles setuptools 63.4.1, with which py37 gate\nworks well.\n\nChange-Id: I467f8105ebf13557ee1fb539d3d36e4d1e968118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/39b9dcc20fc2bf785ac4cc393f12ea849bc2670f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/39b9dcc20fc2bf785ac4cc393f12ea849bc2670f"}]},"branch":"refs/heads/stable/ussuri"},"894b912d828c7746bf8c4ce4f2416500ad0f9b1a":{"kind":"REWORK","_number":2,"created":"2022-09-14 18:02:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/855946/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/46/855946/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/46/855946/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/46/855946/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/46/855946/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/46/855946/2"}}},"commit":{"parents":[{"commit":"2ca70cec88f789fbcbcc9075c2253e30d14f3288","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2ca70cec88f789fbcbcc9075c2253e30d14f3288"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-05 15:08:25.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-14 18:02:00.000000000","tz":60},"subject":"[stable-only] Cap virtualenv for py37","message":"[stable-only] Cap virtualenv for py37\n\npy37 test started to fail with latest virtualenv (20.16.4) as it\nbundles setuptools 65.3.0, which causes installation problems under\npy37. (Note: with py38 the problem does not occur).\n\nThis patch sets virtualenv\u003c20.16.4 to unblock the gate as\nvirtualenv 20.16.3 bundles setuptools 63.4.1, with which py37 gate\nworks well.\n\nNOTE(stephenfin): Modified to remove the (failing) l-c job.\n\nChange-Id: I467f8105ebf13557ee1fb539d3d36e4d1e968118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/894b912d828c7746bf8c4ce4f2416500ad0f9b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/894b912d828c7746bf8c4ce4f2416500ad0f9b1a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
