)]}'
{"id":"openstack%2Foslo.middleware~874210","triplet_id":"openstack%2Foslo.middleware~master~I303f753e47eade9d5de9dad0b3d9cc303f6a0540","project":"openstack/oslo.middleware","branch":"master","topic":"oslo-tox-4","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-17 18:56:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I303f753e47eade9d5de9dad0b3d9cc303f6a0540","subject":"Fix issues related to tox4","status":"MERGED","created":"2023-02-17 09:24:26.000000000","updated":"2023-02-17 18:57:20.000000000","submitted":"2023-02-17 18:56:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874210-oslo-tox-4","meta_rev_id":"2c099aeea1b0ef8dea97cb38c9a49bd3b10f4160","_number":874210,"virtual_id_number":874210,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 18:56:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-02-17 15:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-02-17 15:34:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 11:24:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 15:29:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"c22556e09f813a5bebe0536ea433879399eb1c85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 09:24:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d5918541c9a7e9843797e689fe6b7e9b40179b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:24:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b97bbb717a524fdca5e03c9675cc8146\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/153453d9d53e40d1858c462e7d5fddb5 : SUCCESS in 1h 54m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bfcf222c1b4421e9b804a78d71f80ac : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/505b44807c574f9ead99c63465d15392 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/772e72cf47894fbeada0bdc32440ba00 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6e8fd0119b401d912ce7d1fdba9338 : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/038e15d91285427e82b3de414d72b892 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fb199c508e7abb0ab9def7b337af29f3f233bcad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 13:16:13.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"08dc2e2aa69069fefeab8f598b8df615c5cd23fe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 13:16:46.000000000","message":"Topic oslo-tox-4 removed","accounts_in_message":[],"_revision_number":2},{"id":"1d96fa81140228fd7b630db5592156451133d4e9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 13:16:51.000000000","message":"Topic set to release-health-check-oslo","accounts_in_message":[],"_revision_number":2},{"id":"ec97434e5a90863d97c18dac55d98d006913dcb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:44:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e59df44b45a4376804b4b6dd8a2a540\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66538954a66340c5894e7545917817fb : SUCCESS in 1h 25m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49d3a13d2be64feca8cb264b3e05a8d8 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fad10613e60447cb6968135519db1de : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2129e24956a241628fdc83a0d98b7986 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a767cbcbd8843cebf6090ac066b07d5 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/707d4dacc5ec4aef95705a76c4d0e017 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ce004369d78efa6314eda1f2d9d610a81cacaa58","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-17 15:29:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"32063c969fecd27a4b61ab595b0f106711e7b322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 15:31:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"59bb2a0b13c20173d56e9128bdbee39755ecadf5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-17 15:34:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8f06e2062d3144d4a7a3e2a6f4e3f1258a3ed8cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 17:20:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b02f5c1f0e44df9a09effc3731519d7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9c27a0e6dd94e9f9c2b39f973d2022a : SUCCESS in 1h 46m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f83dff784a8b49e4a3ffde3da4a17d58 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa62a364f7b843bbb280c77fb6bdfef1 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e3eeac745a34a48917a4fcf38a486d6 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd866d0afc84521ade2825b9e1d88f0 : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ff676008f54457d95b5ec6d3d6d0c20 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":3},{"id":"8feb4a7734ea168e30fc32924e0e02e6c4b57bf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 17:20:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"749d0b4beb581d7f90011090e936ced13a13f64b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 18:56:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2d501947c864c109a3d129e6f4b5ead\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1073154da7414de0ae5858ee6680b1d0 : SUCCESS in 1h 34m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13af100b1a248558fc909309ae908a3 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15f305791468425fb854b0a87b85f338 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b361c165351423bb4a9067b60a9d6f5 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea7b7733b6f411ead207e0a9edbcfcf : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce7dd72fe14e48368f3479038295cfb8 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":3},{"id":"7218be970621d390d4b88dfc899c5332919306b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 18:56:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2c099aeea1b0ef8dea97cb38c9a49bd3b10f4160","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 18:57:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c15cf65f87954a488696a70851675b4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e7508db9d73461d8a01f5f8c19c3b11 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86316f4796b54c2893d65b4f5f2f50d9 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7725ac941fdf24f8686a350906ca493e720a7df3","revisions":{"e3f8488d6c1f7c233ee6db6cba70db41692eef79":{"kind":"REWORK","_number":1,"created":"2023-02-17 09:24:26.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/874210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/10/874210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/1"}}},"commit":{"parents":[{"commit":"c9bd063ebc652ef519e6bd81f4ad5c9a8cbe62ff","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c9bd063ebc652ef519e6bd81f4ad5c9a8cbe62ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 09:24:16.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 09:24:25.000000000","tz":60},"subject":"Fix issues related to tox4","message":"Fix issues related to tox4\n\nChange-Id: I303f753e47eade9d5de9dad0b3d9cc303f6a0540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e3f8488d6c1f7c233ee6db6cba70db41692eef79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e3f8488d6c1f7c233ee6db6cba70db41692eef79"}]},"branch":"refs/heads/master"},"8353e7468aecf522b824ef8045b06b37e0ceaa9f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-17 13:16:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/874210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/10/874210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/2"}}},"commit":{"parents":[{"commit":"c9bd063ebc652ef519e6bd81f4ad5c9a8cbe62ff","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c9bd063ebc652ef519e6bd81f4ad5c9a8cbe62ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 09:24:16.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-17 13:16:13.000000000","tz":0},"subject":"Fix issues related to tox4","message":"Fix issues related to tox4\n\nRelated to https://lists.openstack.org/pipermail/openstack-discuss/2023-February/032247.html\n\nChange-Id: I303f753e47eade9d5de9dad0b3d9cc303f6a0540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/8353e7468aecf522b824ef8045b06b37e0ceaa9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/8353e7468aecf522b824ef8045b06b37e0ceaa9f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7725ac941fdf24f8686a350906ca493e720a7df3":{"kind":"REWORK","_number":3,"created":"2023-02-17 15:31:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/874210/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/10/874210/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/10/874210/3"}}},"commit":{"parents":[{"commit":"c9bd063ebc652ef519e6bd81f4ad5c9a8cbe62ff","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c9bd063ebc652ef519e6bd81f4ad5c9a8cbe62ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 09:24:16.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 15:31:41.000000000","tz":60},"subject":"Fix issues related to tox4","message":"Fix issues related to tox4\n\nChange-Id: I303f753e47eade9d5de9dad0b3d9cc303f6a0540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7725ac941fdf24f8686a350906ca493e720a7df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7725ac941fdf24f8686a350906ca493e720a7df3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
