)]}'
{"id":"openstack%2Foslo.middleware~879287","triplet_id":"openstack%2Foslo.middleware~master~I45f418ea67bd24a91ec0ca1989b7c54b234a2242","project":"openstack/oslo.middleware","branch":"master","topic":"oslo-drop-py38","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-11 11:40:18.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-04-11 11:40:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I45f418ea67bd24a91ec0ca1989b7c54b234a2242","subject":"Moves supported python runtimes from version 3.8 to 3.10","status":"MERGED","created":"2023-04-03 09:11:06.000000000","updated":"2023-05-03 20:03:35.000000000","submitted":"2023-04-11 11:40:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879287-oslo-drop-py38","meta_rev_id":"8ea2ddf74d85fda471cc77ca1438162290b55555","_number":879287,"virtual_id_number":879287,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-11 11:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-04-11 10:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-04-11 10:00:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 10:32:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-10 01:46:23.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-04-11 10:00:46.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f7a2e8d6402b3eea74ecf9d14615fc3bdce1cbe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-03 09:11:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0454610f60b03e2151884e97410613cc0de599e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 10:32:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49a31692674d481b9fbd5560723450b0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/85df62a0706246bba9eac0a3e56cfda5 : SUCCESS in 1m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ccb931051a84066ab90f955678b5f76 : SUCCESS in 1h 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15eb4bb2ff8f4c75966675cc4dd4e0d2 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d7ce563394d4dc08d782b978bfc85c6 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d955a9bdf33a4cb7b6e3e5809c3b3ede : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd60893a645a49fe987a22d9bbb6c9e0 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"71ae68e87bb5ff57befb530f94cf2f149bf5fd5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 01:46:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af6d76d20d78d45145b9743d6fd4a03de128111b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 01:54:17.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7cf1afecafc4bc4d5092d445f3bdcf427f1f2041","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 01:54:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fa0fa3ae367a9963d93253fc846911ba27401aec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 03:29:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c341e5a04b254bb0bac9167009af0095\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3395ec0d06144f05bd9e5a23fa3ef599 : SUCCESS in 1m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5a579d33af54f3eac96cc7811ab19aa : SUCCESS in 1h 33m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aba852187b64b95b6c16d5816fb1199 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2861ef200b0d48fe8cb373cff63c1240 : FAILURE in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/431378b4029b4112884486ad4d371b8c : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1037b6124d24d028618297c9a70e8fd : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"9e9055b121a51e8dafc37b02ea48e1e9e2dca525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 03:59:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"93cb75742bf0cfc11653b51f1ba54c3fa1c86754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 05:04:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/629cf6f0512b48ad9d043fd56c5e5937\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58895cfd78554e828bc0a82548bb313d : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3dab5be9d444925b1b01408a7ccf1cf : SUCCESS in 1h 04m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25108a3c8750465ebac7c2e3042029ff : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/401a77a201c246ecac3436dab7e48075 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7added3092c5420082598d7a9ee99299 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1142bb75fc417499f9c148987324f8 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":3},{"id":"b03913d277846dbf78dfa8d29aba93a7768794f1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-11 10:00:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f2aed9c800a8e21d7b88147a388cff62ec6914bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 10:01:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5878dd88478c8e649645864628b3f61675a99597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 11:40:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38439af78ddf4cbe81378a90b2db60cc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2aeb542c7c51471e8d54fc9563a9511f : SUCCESS in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05e369a188b6443dbbd563abedb24008 : SUCCESS in 1h 29m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fb6bd4941224823b79c4015044be3da : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9edcb38c94fd44e2a28056fe3003d80b : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6fefd5c66904436aaf4263b94494e31 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02547cad9d5a440584a3a2f57225355a : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"afd8fe1e883e9fd4105de999ef26628665c0ceeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 11:40:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"03087f0dfa87e5efb9072986e14224c35662f8d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 11:42:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d27c48bdc744069e21b84f2fb32940\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d385474190d44c42963aa27742e6ef3b : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3},{"id":"8ea2ddf74d85fda471cc77ca1438162290b55555","tag":"autogenerated:gerrit:revert","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-03 20:03:35.000000000","message":"Created a revert of this change as I3a9f8ee9ac1fabbc83680051d5930e956ba113f8","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"089fa3659ee31e08d5adc63f50bcf2bb8c1c7bd2","revisions":{"fb61693e784abf1499d96ba1983aac50583eef91":{"kind":"REWORK","_number":1,"created":"2023-04-03 09:11:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/879287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/87/879287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/1"}}},"commit":{"parents":[{"commit":"05be604d8b19c9db6f8c04333b4ff8a8a556e691","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/05be604d8b19c9db6f8c04333b4ff8a8a556e691"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 09:10:53.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 09:10:53.000000000","tz":120},"subject":"Moves supported python runtimes from version 3.8 to 3.10","message":"Moves supported python runtimes from version 3.8 to 3.10\n\nWithin 2023.2 python version 3.9 and 3.10 are the\nsupported python runtimes [1].\n[1] https: //review.opendev.org/c/openstack/governance/+/872232\n\nChange-Id: I45f418ea67bd24a91ec0ca1989b7c54b234a2242\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/fb61693e784abf1499d96ba1983aac50583eef91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/fb61693e784abf1499d96ba1983aac50583eef91"}]},"branch":"refs/heads/master"},"cba13268e85e83a12ab8bcbf8826ae65174e5e13":{"kind":"REWORK","_number":2,"created":"2023-04-10 01:54:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/879287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/87/879287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/2"}}},"commit":{"parents":[{"commit":"05be604d8b19c9db6f8c04333b4ff8a8a556e691","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/05be604d8b19c9db6f8c04333b4ff8a8a556e691"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 09:10:53.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-10 01:54:16.000000000","tz":0},"subject":"Moves supported python runtimes from version 3.8 to 3.10","message":"Moves supported python runtimes from version 3.8 to 3.10\n\nWithin 2023.2 python version 3.9 and 3.10 are the\nsupported python runtimes [1].\n[1] https: //review.opendev.org/c/openstack/governance/+/872232\n\nChange-Id: I45f418ea67bd24a91ec0ca1989b7c54b234a2242\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/cba13268e85e83a12ab8bcbf8826ae65174e5e13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/cba13268e85e83a12ab8bcbf8826ae65174e5e13"}]},"branch":"refs/heads/master"},"089fa3659ee31e08d5adc63f50bcf2bb8c1c7bd2":{"kind":"REWORK","_number":3,"created":"2023-04-10 03:59:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/879287/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/87/879287/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/87/879287/3"}}},"commit":{"parents":[{"commit":"05be604d8b19c9db6f8c04333b4ff8a8a556e691","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/05be604d8b19c9db6f8c04333b4ff8a8a556e691"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 09:10:53.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-10 03:59:22.000000000","tz":540},"subject":"Moves supported python runtimes from version 3.8 to 3.10","message":"Moves supported python runtimes from version 3.8 to 3.10\n\nWithin 2023.2 python version 3.9 and 3.10 are the\nsupported python runtimes [1].\n[1] https: //review.opendev.org/c/openstack/governance/+/872232\n\nChange-Id: I45f418ea67bd24a91ec0ca1989b7c54b234a2242\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/089fa3659ee31e08d5adc63f50bcf2bb8c1c7bd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/089fa3659ee31e08d5adc63f50bcf2bb8c1c7bd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
