)]}'
{"id":"openstack%2Foslo.middleware~904216","triplet_id":"openstack%2Foslo.middleware~master~Icf2296c1ad1548ffe8be0610800f1c6c5d33c473","project":"openstack/oslo.middleware","branch":"master","topic":"pre-commit","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-03-30 14:35:48.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-30 14:35:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf2296c1ad1548ffe8be0610800f1c6c5d33c473","subject":"pre-commit: Integrate bandit","status":"MERGED","created":"2023-12-21 15:37:15.000000000","updated":"2024-03-30 14:36:50.000000000","submitted":"2024-03-30 14:35:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904216-pre-commit","meta_rev_id":"77f73b7520c1e1835ed5a5520f1585fb2eac114d","_number":904216,"virtual_id_number":904216,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-30 14:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-03-29 01:35:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-29 03:23:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-03-29 01:35:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-21 17:13:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-07 07:23:23.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-03-29 03:23:42.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"d506a18e49053e71f417fac085e0225b844a7280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-12-21 15:37:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"038a1a2add418a3da30a9d4a5f487874c7dd44ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 17:13:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/273f11604a8b4182b47062bc9a808849\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1f685879e5ae4c1c8ebb19784b82e12e : SUCCESS in 1m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0bbdf21e368641e39cf372b5f1b1ce30 : SUCCESS in 1m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15b78801bd2c4affb58e4d1e7fa68ee0 : SUCCESS in 1h 26m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a057dcd378b341c6a757d548b58efc37 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49ed836c523b4547b37ca721bbc761d2 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01ca18809e834adebfb55b6081114f70 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/190bd2090b4044c795151093a550c875 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fdd085d3a9b4011bc05e2beea4bd053 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e83e138e1faa64ee1e448dea65c8a2bbf25cce74","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 07:23:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3124db4dd271e19f1df74bcba0a37fbba6cee3bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-29 01:34:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3e7d66129d5c5789d96ef475406669283a49963c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-29 01:35:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a290c1c1402bdb3448490630c4cfc02aabb0bff2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-29 02:10:32.000000000","message":"Topic typing removed","accounts_in_message":[],"_revision_number":2},{"id":"24ec58ec55094b711272a765f68d2667876fd315","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-29 02:10:37.000000000","message":"Topic set to pre-commit","accounts_in_message":[],"_revision_number":2},{"id":"4f66ba06e7a6577d1704cac80f19c2e23a2f7aea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 02:52:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8760d1246a3043a3a23e1d0c51c93bcb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/518589abc76a4708b0055afe8461df1c : SUCCESS in 1m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b996b52d38f4fecb92f94b6d363a35d : SUCCESS in 1m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2d0e93cecf341008319f96a011dcb62 : SUCCESS in 1h 16m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e941e5834ae4208ae99845a21230698 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f91ba1f022f549a4a3f12484ebdf4304 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/696dc46ed0ca4cc39d5b2f80ba3e26a3 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d579af58719b4cb689f31181d9308d48 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff83a9992a50461984ceb6a626ae688d : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71b96b09dff34648856f28bac83d3642 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"10ea8890be5d59329fa1ea19b5e54451169c7d6b","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2024-03-29 03:23:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fab2ec0ed92683d46f813f34db395d69fcc08487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 12:45:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3062706a670bca6cf9086f27e0e5a9693bbd3d0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 14:35:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60c50365591c4a60b8717a7951da0e08\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/73ff21b0d4b14ffe9589827e1076290b : SUCCESS in 3m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/823e4dd7e9b34d16a4b3a2e340ade3f0 : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02287e405f354e4fb490aec1f01934d9 : SUCCESS in 1h 42m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f7e23005c3c4635bd7036df114bdb7b : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4ae991889734464aa4afa76c2ec690e : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc87fcdba8614e468df786140c0f735b : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88b6c6bf24fe4d43b8df816fc3c931d7 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca2c59f816934333be567fcbac6c3bd5 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"c839ddd4d920279505ef50ffd902beee1865899a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 14:35:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"77f73b7520c1e1835ed5a5520f1585fb2eac114d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 14:36:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd2c87b4bce347ee9b24561c1869882a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0deb5350edf64e0d90ab6c0eea155074 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/490eb7fb7c1e412fa5b83a47728a3b75 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"842e5d047cf2e6ced7de2454b6644a580ec806f2","revisions":{"00a141d440b27042174d687138331a56882c52b5":{"kind":"REWORK","_number":1,"created":"2023-12-21 15:37:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/904216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/16/904216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/16/904216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/16/904216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/16/904216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/16/904216/1"}}},"commit":{"parents":[{"commit":"2e9a5bea0daa9596fce52344aa3416bac582b32a","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2e9a5bea0daa9596fce52344aa3416bac582b32a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-21 15:32:59.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-21 15:35:50.000000000","tz":0},"subject":"pre-commit: Integrate bandit","message":"pre-commit: Integrate bandit\n\nWe also remove these unnecessary linter dependencies from\ntest-requirements.txt.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Icf2296c1ad1548ffe8be0610800f1c6c5d33c473\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/00a141d440b27042174d687138331a56882c52b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/00a141d440b27042174d687138331a56882c52b5"}]},"branch":"refs/heads/master"},"842e5d047cf2e6ced7de2454b6644a580ec806f2":{"kind":"REWORK","_number":2,"created":"2024-03-29 01:34:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/904216/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/16/904216/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/16/904216/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/16/904216/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/16/904216/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/16/904216/2"}}},"commit":{"parents":[{"commit":"c2a6d48828b5245bba371267c488592639cbac20","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c2a6d48828b5245bba371267c488592639cbac20"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-21 15:32:59.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-29 01:34:19.000000000","tz":540},"subject":"pre-commit: Integrate bandit","message":"pre-commit: Integrate bandit\n\nWe also remove these unnecessary linter dependencies from\ntest-requirements.txt.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Icf2296c1ad1548ffe8be0610800f1c6c5d33c473\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/842e5d047cf2e6ced7de2454b6644a580ec806f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/842e5d047cf2e6ced7de2454b6644a580ec806f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
