)]}'
{"id":"openstack%2Foslo.middleware~920055","triplet_id":"openstack%2Foslo.middleware~master~Ia8a54cbd38f2faf70bbe91032da36448f2fa2de7","project":"openstack/oslo.middleware","branch":"master","topic":"healthcheck/enable_by_files","attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-04 09:38:54.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-04 02:51:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-07-04 04:24:04.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-07-04 04:24:04.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-07-03 16:00:30.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"Ia8a54cbd38f2faf70bbe91032da36448f2fa2de7","subject":"Add a enabled by files healthcheck plugin","status":"MERGED","created":"2024-05-20 16:07:28.000000000","updated":"2024-07-04 09:38:54.000000000","submitted":"2024-07-04 04:24:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"920055-healthcheck/enable_by_files","meta_rev_id":"6ae63648301be8efd4208fcdfa32995f4ed46de4","_number":920055,"virtual_id_number":920055,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 04:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-03 14:07:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-07-04 02:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-07-03 16:00:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-07-04 02:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-20 16:09:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-05-20 16:09:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-05-20 16:09:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-05-20 17:38:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-02 09:48:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"075cacd53e62e4b4f140859b2b8c102b4aea9cf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-20 16:07:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af0ef40f89549d665bd3022aeb74498a2f2af505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 17:38:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be81dba025d7446e9113d51016d8965c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ef71a0ab29449c8a6fd2cab7717c5ef : SUCCESS in 1m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c853a6bd139847cba13c730fc258a872 : SUCCESS in 1h 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4667552fbe77474799c3ff4bed8150e5 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/107c9b9e4fec4b5c9fb8772b46ac3366 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02ef23a239d8484cb7e9c72012870109 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8dd896e071834df3b0dcc4096290bd42 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab610a51065430999b689d15ac1e016 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23e1638124304029b3d90ad1cd0c495f : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9b51039b759d7bba23fd41ac15110dc00a95e721","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-20 16:28:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"143fdcf0559446c18bf5a319be5c58500d0149b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 06:56:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"082bf7ded601c1e9ab5d648e36b829ae181d8b04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 06:58:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1763ffbe969ac73d0845ae12f308a1dd0080b9dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 08:43:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b1819e1e1dd444a80488dea12c0bcd3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b23ce2c6d4544b32b4f8ca109b119056 : SUCCESS in 3m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ace6e287bac4714adaf33a9966ad317 : SUCCESS in 1h 42m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f9edd99daa64b5b9ec1e2199a16c08a : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67ecfcff5ec048f39ffbe0adb78b1fbd : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8c7388070ba458eac0ebebdc70fafbb : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90cb78d9870244d09a4bd72efeb641d2 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca99d0771db42de85620672a14f0e68 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f0d8b01febb4248aa15bc2984b379ee : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d90ad0f725776195697213bea123daf4dec86149","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-07-02 09:48:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"009d0e0a64148d5851672916c085396e4aaf8933","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 09:55:46.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"571642ff9208547232d2c8cd0cde6be45886a4cc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-02 13:52:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5dc119fa1c71fc7caefefc7c55d451f94085506","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-07-02 15:11:09.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a793b5855332bbba3da78282b9f93157752240b6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 15:15:22.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"734cd16dd94fbb6d99dadc681633e8d8d9b48edf","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-02 21:12:39.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"61d0e8bbbee4d50a0e6d92e2ccbcb36352e7dabe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-02 21:13:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a414d406327708ad11b3ac27e21075be1db7fe8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-02 21:14:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fb04adb200b90dc63ad2670ccdd329c4c4d548f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 22:59:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b1ad92e68fe47c08fb567f95b0c1ee9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/88f91f3d77374789b8bf7bed1efa9013 : SUCCESS in 4m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29e167ee9e7d44ddaa44099de3d835c1 : SUCCESS in 1h 42m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21615886607b4a329f2ba71cbe5bda96 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30f826fd06e74e2a8b7d0cd13b233d3e : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fab4450c0a64c28a7829507895143ac : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd8bfd71f11f4cc6932bab566a5e4b3b : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/955a236024f9450cbdb976d5f2dae0b6 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cda65a01b5a749babf66acff3b05c2d1 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":5},{"id":"a3c83ee761bed4d891dd2298e2cbdeca5fb630d2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 02:02:58.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c1ae4950288901420e53c1e97403ab254971022e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-03 13:55:31.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"78f9c1156de5405be13121ec1fa58e6505343828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-03 13:55:42.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e707a1906bf4a7b960de634b394a46687d83162a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:07:24.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7c6b9b6feaa0645a7db0522f80f54f55a3ab4d54","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-07-03 16:00:30.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b79ca15a7f75e06278dfc1a29a98799f7550cfb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 16:45:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08be81c94ef3478ebe912f011dce4b99\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7a01116bf9bd4619bef41572c22cc7b7 : SUCCESS in 1m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/265f4e6819cb4dcda7945cb8c3ae61fd : SUCCESS in 1h 38m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8eb378e7ea41458adb668715b52177 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcbe11db7b244b60b073dfdc5ac54f39 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb3d28c979764f2fad032cb97cb6d0b7 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/116ee2f14065436b9f4ffac39ad7f496 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a24286f9d3e24cda95b84025ac0cf42e : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc15a93382c8496384f3cfbba4f533e7 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":6},{"id":"b38d6ec23f4367425706fd186d7ac6df2e2d70ab","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-04 02:51:02.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6b47cf0cf84d17333f16d41588fe7a9389730301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 02:51:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4bb682cbb2db95389ee62ebe423c2eb867313d17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 04:24:04.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbf32481167e4bac9473ccec4e2650db\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a12944e38c2f41f6988f741bbc79b43f : SUCCESS in 1m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0721970a4ccc4347ae9091d752a660b0 : SUCCESS in 1h 30m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a11093be85e446a85ef18f162e2f2f9 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1a6cadd698043609ba3da2b26789ccd : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5d9dbd78cb546bbafec4016dfa6f79d : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0a6b1a495344af4978212a963fc8cf6 : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7871dddb44c34867baf43b7a2db50cf3 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":6},{"id":"438b9465082643541005665c97b158321bba2a99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 04:24:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6469d9b9170832d0243048c86ea7b36cb3668bfd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 04:25:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efafb96f655f4663a931551643237a28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31817cc42db34f0ba956de9712e07a56 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1df9d014de6e497d939b7d82659be4fb : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":6},{"id":"6ae63648301be8efd4208fcdfa32995f4ed46de4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-07-04 09:38:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"01cd6081044256aaaafe8cf149e4e91801bf66b7","revisions":{"87148d08faddbc4c2accd5b2711023b720aa9251":{"kind":"REWORK","_number":1,"created":"2024-05-20 16:07:28.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/55/920055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/55/920055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/1"}}},"commit":{"parents":[{"commit":"e2ea8dc556a7e8944b923ffd2a0028495bb60b5e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e2ea8dc556a7e8944b923ffd2a0028495bb60b5e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-05-20 16:04:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-05-20 16:07:15.000000000","tz":0},"subject":"Add a enabled by files healthcheck plugin","message":"Add a enabled by files healthcheck plugin\n\nImplements: blueprint enable-by-files-healthcheck\nChange-Id: Ia8a54cbd38f2faf70bbe91032da36448f2fa2de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/87148d08faddbc4c2accd5b2711023b720aa9251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/87148d08faddbc4c2accd5b2711023b720aa9251"}]},"branch":"refs/heads/master"},"e16a5158df33138927c424bacd84cd5539c56e02":{"kind":"REWORK","_number":2,"created":"2024-07-02 06:56:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/55/920055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/55/920055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/2"}}},"commit":{"parents":[{"commit":"e2ea8dc556a7e8944b923ffd2a0028495bb60b5e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e2ea8dc556a7e8944b923ffd2a0028495bb60b5e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-05-20 16:04:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-02 06:56:05.000000000","tz":0},"subject":"Add a enabled by files healthcheck plugin","message":"Add a enabled by files healthcheck plugin\n\nImplements: blueprint enable-by-files-healthcheck\nChange-Id: Ia8a54cbd38f2faf70bbe91032da36448f2fa2de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e16a5158df33138927c424bacd84cd5539c56e02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e16a5158df33138927c424bacd84cd5539c56e02"}]},"branch":"refs/heads/master"},"d9f2db135dfef7f27f44d64cf07b78e53f0c8a29":{"kind":"REWORK","_number":3,"created":"2024-07-02 06:58:30.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/55/920055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/55/920055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/3"}}},"commit":{"parents":[{"commit":"e2ea8dc556a7e8944b923ffd2a0028495bb60b5e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e2ea8dc556a7e8944b923ffd2a0028495bb60b5e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-05-20 16:04:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-02 06:57:35.000000000","tz":0},"subject":"Add a enabled by files healthcheck plugin","message":"Add a enabled by files healthcheck plugin\n\nImplements: blueprint enable-by-files-healthcheck\nChange-Id: Ia8a54cbd38f2faf70bbe91032da36448f2fa2de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d9f2db135dfef7f27f44d64cf07b78e53f0c8a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d9f2db135dfef7f27f44d64cf07b78e53f0c8a29"}]},"branch":"refs/heads/master"},"0340295692429a84abd3b8166657caa1c7929236":{"kind":"REWORK","_number":4,"created":"2024-07-02 21:13:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/55/920055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/55/920055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/4"}}},"commit":{"parents":[{"commit":"e2ea8dc556a7e8944b923ffd2a0028495bb60b5e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e2ea8dc556a7e8944b923ffd2a0028495bb60b5e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-05-20 16:04:36.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-02 21:12:53.000000000","tz":120},"subject":"Add a enabled by files healthcheck plugin","message":"Add a enabled by files healthcheck plugin\n\nImplements: blueprint enable-by-files-healthcheck\nChange-Id: Ia8a54cbd38f2faf70bbe91032da36448f2fa2de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/0340295692429a84abd3b8166657caa1c7929236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/0340295692429a84abd3b8166657caa1c7929236"}]},"branch":"refs/heads/master"},"97739a52fa9cc5dd9b416263bf876d1c6f6da978":{"kind":"REWORK","_number":5,"created":"2024-07-02 21:14:52.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/55/920055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/55/920055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/5"}}},"commit":{"parents":[{"commit":"e2ea8dc556a7e8944b923ffd2a0028495bb60b5e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e2ea8dc556a7e8944b923ffd2a0028495bb60b5e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-05-20 16:04:36.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-02 21:14:34.000000000","tz":120},"subject":"Add a enabled by files healthcheck plugin","message":"Add a enabled by files healthcheck plugin\n\nImplements: blueprint enable-by-files-healthcheck\nChange-Id: Ia8a54cbd38f2faf70bbe91032da36448f2fa2de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/97739a52fa9cc5dd9b416263bf876d1c6f6da978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/97739a52fa9cc5dd9b416263bf876d1c6f6da978"}]},"branch":"refs/heads/master"},"01cd6081044256aaaafe8cf149e4e91801bf66b7":{"kind":"REWORK","_number":6,"created":"2024-07-03 13:55:42.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/55/920055/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/55/920055/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/55/920055/6"}}},"commit":{"parents":[{"commit":"e2ea8dc556a7e8944b923ffd2a0028495bb60b5e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e2ea8dc556a7e8944b923ffd2a0028495bb60b5e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-05-20 16:04:36.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-03 13:55:15.000000000","tz":120},"subject":"Add a enabled by files healthcheck plugin","message":"Add a enabled by files healthcheck plugin\n\nImplements: blueprint enable-by-files-healthcheck\nChange-Id: Ia8a54cbd38f2faf70bbe91032da36448f2fa2de7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/01cd6081044256aaaafe8cf149e4e91801bf66b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/01cd6081044256aaaafe8cf149e4e91801bf66b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
