)]}'
{"id":"openstack%2Foslo.middleware~969264","triplet_id":"openstack%2Foslo.middleware~master~I17d8e2d78d95c0d780dbefe084092738962016f0","project":"openstack/oslo.middleware","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-05 17:24:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-12-05 18:31:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I17d8e2d78d95c0d780dbefe084092738962016f0","subject":"Run mypy from tox","status":"MERGED","created":"2025-12-02 15:35:19.000000000","updated":"2025-12-05 18:32:13.000000000","submitted":"2025-12-05 18:31:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969264-typing","meta_rev_id":"cc5010f5e25d69e44c472eda1ac0b613cd15dbee","_number":969264,"virtual_id_number":969264,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 18:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-05 17:24:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-05 17:24:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-02 16:32:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-03 10:18:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"e974dcbe58203c3993cade9d3a3717c717ceb4b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-02 15:35:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f34311e61c74e1bca3b820ff88c9d896fd221814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 16:32:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58edbbccd808441d93611df693a9ce11\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ffe894bd5f2d40628c4aebf81a57e6dc : SUCCESS in 1m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc97131c5a5c46b0a17cc12d92f8c082 : SUCCESS in 1m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e94c80164c8443aaa0a56eb7202eb5fb : SUCCESS in 55m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea7dc3140a94037b1cb70c04aefb08e : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4fc22f7a1b4a403eb663da0d4fc4b448 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e5875f3f66f4c089d0440b355409e34 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cbfa20a0efb41c7af8b2cdb4f14b46d : SUCCESS in 1m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4176c0346b34c2a9a898e0a24f86671 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d6794219542d2d93ca1e696c05790f5358d5647f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-12-03 10:18:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6bb1e08555d4836ab3b0c326de13cd00456b622","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-04 18:16:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d81503f7babf53b2a2da1849d8f7cd2f138b7755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 20:02:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa234df201564615813492fae6a4a5a8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bf31ec222c1a429f9b319add7b383253 : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/54ef153193224d7eaa745f2af5e7b84f : SUCCESS in 1m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efc1992951ff49f38f4ff28118386034 : SUCCESS in 1h 39m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf69f6acdc4e43398f403427b5f02ec5 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d6a1d5f227e45929323cd2ec20c6c67 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd02b4fb55634b4486dbee2fe181bb5b : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49c3f98b72c94bcf9da5bbc108f890c8 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8b48c70d13d4531afa803cba4dfcb9d : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":2},{"id":"cf3c362f48489b34ccd6b95a327426b5bccf5a1e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-05 17:24:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9da645950854937c5bb8b45709774ccc843309cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:25:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f5d2abdc5fbd21aedba49808bca01969ee12c7ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:31:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99e563ea4deb4cfa8d410d65622e6b66\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/00fb21bf0a124e5fb49e5967ac4f62de : SUCCESS in 1m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/705d794a0b854377b08b96b96d967cc7 : SUCCESS in 1m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8510665e5faf4ae7b87aba97cde3c5fa : SUCCESS in 56m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ed236fd8824f2e9a58b1a1983d41bb : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3dff4d83cd84fe28b50346422c66173 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86407fe857c5424cb060291c25bac575 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1788efe5c1fe49989b4f121806cfb362 : SUCCESS in 1m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfd4f162d1b141dd9c88109698111981 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8263175c5f77cf16c9644baf37375b04df3bde56","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:31:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cc5010f5e25d69e44c472eda1ac0b613cd15dbee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:32:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6db2f7283c64421ae33eef04bb49236\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c30efcb63e4e04a2c54fa5ecdd4fcf : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adaf9c102833404b873b8acab6d770bf : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e050ac6947b3b62800f40123307ead5c749091a2","revisions":{"f047df6d393db511ef3d8a7d3274f12269f09c7c":{"kind":"REWORK","_number":1,"created":"2025-12-02 15:35:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/969264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/64/969264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/64/969264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/64/969264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/64/969264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/64/969264/1"}}},"commit":{"parents":[{"commit":"c83131ed3e185f667bed36812ddb615fe2b29918","subject":"Merge \"cors: Remove multiple config block parsing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c83131ed3e185f667bed36812ddb615fe2b29918"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-02 15:26:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-02 15:35:09.000000000","tz":0},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple\nplaces and allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints.\n\nNote that the change brings in stevedore type hints, which necessitates\nsome small fixes.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I17d8e2d78d95c0d780dbefe084092738962016f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/f047df6d393db511ef3d8a7d3274f12269f09c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/f047df6d393db511ef3d8a7d3274f12269f09c7c"}]},"branch":"refs/heads/master"},"e050ac6947b3b62800f40123307ead5c749091a2":{"kind":"REWORK","_number":2,"created":"2025-12-04 18:16:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/969264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/64/969264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/64/969264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/64/969264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/64/969264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/64/969264/2"}}},"commit":{"parents":[{"commit":"c83131ed3e185f667bed36812ddb615fe2b29918","subject":"Merge \"cors: Remove multiple config block parsing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c83131ed3e185f667bed36812ddb615fe2b29918"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-02 15:26:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-04 18:16:19.000000000","tz":0},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple\nplaces and allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints.\n\nNote that the change brings in stevedore type hints, which necessitates\nsome small fixes.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I17d8e2d78d95c0d780dbefe084092738962016f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e050ac6947b3b62800f40123307ead5c749091a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/e050ac6947b3b62800f40123307ead5c749091a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
