)]}'
{"id":"openstack%2Foslo.middleware~979571","triplet_id":"openstack%2Foslo.middleware~master~I86ab63bab0b5f0750d26767e9489abd39454af33","project":"openstack/oslo.middleware","branch":"master","topic":"linter-sync","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-09 14:13:53.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I86ab63bab0b5f0750d26767e9489abd39454af33","subject":"Update packaging configuration","status":"MERGED","created":"2026-03-09 12:23:04.000000000","updated":"2026-05-02 15:42:57.000000000","submitted":"2026-05-02 15:41:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979571-linter-sync","meta_rev_id":"62c5b04e5eb0e94344d7437bfe047da6e022129a","_number":979571,"virtual_id_number":979571,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-02 15:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-05-02 12:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-05-02 12:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 13:53:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cfd5c469e4b03a88338c5105d1cbc5eeb161838e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 12:23:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dadcd4821c2913663c87ff76a089074daff52ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:53:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6a4e824a4b145bfacd3f22006aedd5e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed57a6ac225c46de81cd555e3fc81b7a : FAILURE in 1m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1b0d5aadc3064aa19021ea7a51b20543 : SUCCESS in 1m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0d03fc0754a4bafbd3d369c47977ec9 : SUCCESS in 1h 25m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bafed27cc26249ff80ec6140c0945f6d : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fbba7aa9a4249898e28040af5a57690 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/561a7fcf84154ac1863bb21d21fd98c6 : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f1f54a84e62f4479b64c3308759fd768 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec9b17442fa4fc9b12189b167b1a473 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f88d9c734f2a88ae9a1f265d567540bbadd245da","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 14:13:53.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a08ebfd67f13e42d55c9420735df42e03b309c7b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 12:27:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"686787aca1fdc519a319e5c97ebd1df8fffdc01a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 13:40:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a83ca614f7fe4f6c92154a0e4ba2600b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f22634f84b524b9cab825c43e8997562 : SUCCESS in 2m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3f9ee73e9c78483b99a26196c58b6d71 : SUCCESS in 2m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b30e01e4def54a9680f607c7097f8e0c : SUCCESS in 1h 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951eabd043d44322925f70a04bf92fc5 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bd8c3e0382944b9b629dae88fce07b2 : SUCCESS in 5m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e62c34543a9543fbadd1040f0d36f6de : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03745143e20641fa8084ee3d7417efcd : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c06d066971b04058976dbf70bc9cc910 : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3c8b932a3c1c361a30ccd4275f43d138a55fa6a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 13:40:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b856da7f0640add030b792cc62d212480cc75b7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 15:41:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cab5b93f590e4aecba0608882807f7fe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3c9383ae8d9743d58411a5f12ad8f699 : SUCCESS in 2m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6cebaaa4f944686b1fb7ea286a880b2 : SUCCESS in 2m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce36d5aa2a074f92a6494999f225818f : SUCCESS in 1h 55m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb1975762144ef7a1e23f8a54e1a4ca : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/341fdc8d34ad42f685d89edbd83df46a : SUCCESS in 2m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86975a6e4b944cf997e086884fc1e9ea : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb5ff12bcda486287913d0a654aef3d : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d690e6e0291789a6e3c18cb42adecc598d3b9d56","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 15:41:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"62c5b04e5eb0e94344d7437bfe047da6e022129a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 15:42:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbbc8bf722234f5fb4e46a348620db3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2485bcb93b24579aad8114e5f544fe1 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f233af314b077b62d01d524e92e9f6bb88e3f576","revisions":{"f233af314b077b62d01d524e92e9f6bb88e3f576":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:23:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/979571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/71/979571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/71/979571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/71/979571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/71/979571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/71/979571/1"}}},"commit":{"parents":[{"commit":"2807a8f83022d28202045d5b0c2f498f6982d05d","subject":"ruff: Configure hacking as external linter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2807a8f83022d28202045d5b0c2f498f6982d05d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 12:22:58.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 12:22:58.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I86ab63bab0b5f0750d26767e9489abd39454af33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/f233af314b077b62d01d524e92e9f6bb88e3f576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/f233af314b077b62d01d524e92e9f6bb88e3f576"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
