)]}'
{"id":"openstack%2Foslo.policy~1007385","triplet_id":"openstack%2Foslo.policy~master~I1e32efd9ddd4212cf4420843606843ab5cd09be1","project":"openstack/oslo.policy","branch":"master","topic":"py315","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-28 10:35:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I1e32efd9ddd4212cf4420843606843ab5cd09be1","subject":"Declare Python 3.15 support","status":"MERGED","created":"2026-09-25 04:45:11.000000000","updated":"2026-09-28 12:24:58.000000000","submitted":"2026-09-28 12:21:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007385-py315","meta_rev_id":"41ede47528a5279a4b3417ead7e19eb514dd3105","_number":1007385,"virtual_id_number":1007385,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 12:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-09-27 02:44:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-28 10:35:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 17:03:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-27 02:44:52.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"3fd55d4c03f2a7081624a0e75ef4b8616f3ef497","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 04:45:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8659504651eb9148145b2d337b7516127f3c737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 15:20:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c1707e143e1903a768b6230496c0e03f23696ea8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 15:20:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9a200920292dfdd408d66261d3747988c7841b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 17:03:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b22324c7794a3684ac12eadd4787da\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/559647b277fe4797942d1967008fad9f : SUCCESS in 3m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/127e2f347d5f439994f7b18f7ce5ae07 : SUCCESS in 2m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3184c305cb64eee8816f1c489256b35 : SUCCESS in 1h 34m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f0f34b3c524c819220ea4ed1ccde30 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c27699ecb7744cce9f81a786d5198fa8 : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8e2094792e347b9a869b4b7626f2f7c : SUCCESS in 5m 40s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/40b276ddf6c0492d837e82ef5f00cf14 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d2e6845d1f478286a3c888428e592b : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b8527f5debb4fd59deb5b93c7ce144a : SUCCESS in 3m 27s\n- cross-nova-tox-py314 https://zuul.opendev.org/t/openstack/build/8f7fe4b542a34642a791275e7660cf6d : SUCCESS in 17m 13s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/ae1127d4c4b94613ba04190b90611a63 : SUCCESS in 31m 25s\n- cross-neutron-tox-py314 https://zuul.opendev.org/t/openstack/build/9349ae654b60453a97294cef38450798 : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":3},{"id":"b6287ad19590043d50d65c5a6daa9a111834b6b5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-27 02:44:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"28e33d204314cbff1c1e35de1dbeee82a330b29a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-28 10:35:22.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e8bc7a74041437583c07fffb2bb95300a3af32d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 10:39:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cacb608d629a169776316826237af92bc2f9ba36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:21:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6e2836a70c14342a97421f8a65046c0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dbcc5b311bcf4d95ba18f1863add95fb : SUCCESS in 4m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4ce0a26042af49a1a964024aa99c4379 : SUCCESS in 3m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2322d89ca9674e3da6128c5562e959fe : SUCCESS in 44m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05cbfd8ef13f422d81d1cc600e916d60 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e24d7c270b184798a71a1d37fa597df0 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2fec637d53ef474eadf19be9c2eab027 : SUCCESS in 5m 49s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/bae0d9d88b22403b838a2be69ff06c03 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e660f6be5ad4fe0b9e9db69101304a3 : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3db02a9925e145a5b0987d3a834e62aa : SUCCESS in 5m 52s\n- cross-nova-tox-py314 https://zuul.opendev.org/t/openstack/build/fd0909a7ae0b42f6b1f5ea835143e81a : SUCCESS in 19m 46s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/7c168d45112149e48a1825437029287f : SUCCESS in 28m 36s\n- cross-neutron-tox-py314 https://zuul.opendev.org/t/openstack/build/0ac9f53ba66241ceb4510f033891d08c : SUCCESS in 30m 44s","accounts_in_message":[],"_revision_number":3},{"id":"ade13437249902d21146ca785cd739e9b9f6a055","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:21:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"41ede47528a5279a4b3417ead7e19eb514dd3105","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:24:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8ce75b4cd52407599aaa711c1a712fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa05dec9f3fe428799b8a58812b186c0 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a10ad0e8255221a06e7c05679eb73eee2f5d6516","revisions":{"0e749850ce95b4a36dc4c601b356d59d92f2ebfc":{"kind":"REWORK","_number":1,"created":"2026-09-25 04:45:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/1007385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/85/1007385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/1"}}},"commit":{"parents":[{"commit":"9b4f40b02aae8e800ca3814e7c2de780d4dabec1","subject":"Update jobs according to tested runtimes for 2027.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/9b4f40b02aae8e800ca3814e7c2de780d4dabec1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 04:44:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 04:44:44.000000000","tz":540},"subject":"Declare Python 3.15 support","message":"Declare Python 3.15 support\n\nIt\u0027s now voting in the -next job template.\n\nChange-Id: I1e32efd9ddd4212cf4420843606843ab5cd09be1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/0e749850ce95b4a36dc4c601b356d59d92f2ebfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/0e749850ce95b4a36dc4c601b356d59d92f2ebfc"}]},"branch":"refs/heads/master"},"333c8954bf7bd90a776a581a55bab1db1c0a783e":{"kind":"NO_CHANGE","_number":2,"created":"2026-09-25 15:20:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/1007385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/85/1007385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/2"}}},"commit":{"parents":[{"commit":"e7485b25a12dbea422f44046ec236a5ad07a71ff","subject":"Update jobs according to tested runtimes for 2027.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/e7485b25a12dbea422f44046ec236a5ad07a71ff"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 04:44:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 15:20:10.000000000","tz":0},"subject":"Declare Python 3.15 support","message":"Declare Python 3.15 support\n\nIt\u0027s now voting in the -next job template.\n\nChange-Id: I1e32efd9ddd4212cf4420843606843ab5cd09be1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/333c8954bf7bd90a776a581a55bab1db1c0a783e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/333c8954bf7bd90a776a581a55bab1db1c0a783e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9b4f40b02aae8e800ca3814e7c2de780d4dabec1","ours":"0e749850ce95b4a36dc4c601b356d59d92f2ebfc","theirs":"e7485b25a12dbea422f44046ec236a5ad07a71ff","merge_strategy":"recursive","contains_conflicts":false}},"a10ad0e8255221a06e7c05679eb73eee2f5d6516":{"kind":"NO_CHANGE","_number":3,"created":"2026-09-25 15:20:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/1007385/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/85/1007385/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/85/1007385/3"}}},"commit":{"parents":[{"commit":"83a3a9451e31598ffa11bac3e64749fc8f3090a5","subject":"Update jobs according to tested runtimes for 2027.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/83a3a9451e31598ffa11bac3e64749fc8f3090a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 04:44:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 15:20:51.000000000","tz":0},"subject":"Declare Python 3.15 support","message":"Declare Python 3.15 support\n\nIt\u0027s now voting in the -next job template.\n\nChange-Id: I1e32efd9ddd4212cf4420843606843ab5cd09be1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/a10ad0e8255221a06e7c05679eb73eee2f5d6516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/a10ad0e8255221a06e7c05679eb73eee2f5d6516"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e7485b25a12dbea422f44046ec236a5ad07a71ff","ours":"333c8954bf7bd90a776a581a55bab1db1c0a783e","theirs":"83a3a9451e31598ffa11bac3e64749fc8f3090a5","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
