)]}'
{"id":"openstack%2Foslo.policy~658906","triplet_id":"openstack%2Foslo.policy~master~Idead9b4198c6b05d72bae60dee06e5aebc223822","project":"openstack/oslo.policy","branch":"master","topic":"cap-bandit","hashtags":[],"change_id":"Idead9b4198c6b05d72bae60dee06e5aebc223822","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","status":"MERGED","created":"2019-05-13 19:49:16.000000000","updated":"2019-09-24 15:21:45.000000000","submitted":"2019-05-14 00:39:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658906-1557794343489-cdafc7c2","meta_rev_id":"e551ecd2a4d102e52eb88cabd0d29bb805673048","_number":658906,"virtual_id_number":658906,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-05-14 00:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":2,"date":"2019-05-13 22:29:51.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":1,"date":"2019-05-13 22:29:51.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 22:29:51.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-14 00:39:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5807a2b7cb77a8c3659bacd2117559ec9d2669c4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-13 19:49:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1241f85aa62b2667fe11718ad623061e5cfbafe4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-13 19:52:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c417f114d4718e3bd5c2be41cba2f4cfea36976c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 21:46:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/06/658906/2/check/requirements-check/5625133/ : SUCCESS in 3m 17s\n- tempest-full http://logs.openstack.org/06/658906/2/check/tempest-full/194c181/ : SUCCESS in 1h 47m 30s\n- tempest-full-py3 http://logs.openstack.org/06/658906/2/check/tempest-full-py3/e0ff707/ : SUCCESS in 1h 36m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/658906/2/check/openstack-tox-lower-constraints/f0cdb63/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/06/658906/2/check/openstack-tox-pep8/c000513/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/06/658906/2/check/openstack-tox-py27/f847679/ : SUCCESS in 11m 53s\n- openstack-tox-py36 http://logs.openstack.org/06/658906/2/check/openstack-tox-py36/bbde30c/ : SUCCESS in 12m 51s\n- openstack-tox-py37 http://logs.openstack.org/06/658906/2/check/openstack-tox-py37/71259ad/ : SUCCESS in 15m 35s\n- openstack-tox-docs http://logs.openstack.org/06/658906/2/check/openstack-tox-docs/6c03821/html/ : SUCCESS in 5m 48s\n- build-openstack-releasenotes http://logs.openstack.org/06/658906/2/check/build-openstack-releasenotes/8b1ca89/html/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f8a15d2d986fd7927b642c08a25b7dec2c9492eb","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-05-13 22:29:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a3b8b1bdc9a14c628ef1597147726f2c65fd1f3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 22:30:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c498bc61f17ca70cabfa648638dbd8ccf0290125","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:39:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/06/658906/2/gate/requirements-check/7e5814b/ : SUCCESS in 3m 36s\n- tempest-full http://logs.openstack.org/06/658906/2/gate/tempest-full/077378c/ : SUCCESS in 2h 02m 46s\n- tempest-full-py3 http://logs.openstack.org/06/658906/2/gate/tempest-full-py3/a7fc2f8/ : SUCCESS in 1h 37m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/658906/2/gate/openstack-tox-lower-constraints/480925d/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 http://logs.openstack.org/06/658906/2/gate/openstack-tox-pep8/10ec10c/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/06/658906/2/gate/openstack-tox-py27/7e906a3/ : SUCCESS in 6m 02s\n- openstack-tox-py36 http://logs.openstack.org/06/658906/2/gate/openstack-tox-py36/74514ac/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/06/658906/2/gate/openstack-tox-py37/fb2604f/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/06/658906/2/gate/openstack-tox-docs/709be6c/html/ : SUCCESS in 5m 54s\n- build-openstack-releasenotes http://logs.openstack.org/06/658906/2/gate/build-openstack-releasenotes/563f0f6/html/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5ff1a0c942e7faddfba2f71361b26507036ee3df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:39:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d720b1b76a58c5ff9a6c94952b4b1206b7d0989f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-09-24 15:21:45.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 3a22403be8f0c32581d108a2eb4d0674e9c00e40","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d7ca8a198419896346369311fccedea5a559f66","revisions":{"6578e5eafbfd942cdb27ef7e971e464981771a79":{"kind":"REWORK","_number":1,"created":"2019-05-13 19:49:16.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/06/658906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/06/658906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/06/658906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/06/658906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/06/658906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/06/658906/1"}}},"commit":{"parents":[{"commit":"bda4cea4f77269532bbe66929be2007bde801de5","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/bda4cea4f77269532bbe66929be2007bde801de5"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-13 19:48:30.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-13 19:48:30.000000000","tz":0},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start capping\nit there as well.\n\nChange-Id: Idead9b4198c6b05d72bae60dee06e5aebc223822\nReference: https://github.com/PyCQA/bandit/pull/489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/6578e5eafbfd942cdb27ef7e971e464981771a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/6578e5eafbfd942cdb27ef7e971e464981771a79"}]},"branch":"refs/heads/master"},"1d7ca8a198419896346369311fccedea5a559f66":{"kind":"REWORK","_number":2,"created":"2019-05-13 19:52:15.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/06/658906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/06/658906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/06/658906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/06/658906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/06/658906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/06/658906/2"}}},"commit":{"parents":[{"commit":"bda4cea4f77269532bbe66929be2007bde801de5","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/bda4cea4f77269532bbe66929be2007bde801de5"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-13 19:48:30.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-13 19:52:07.000000000","tz":0},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start capping\nit there as well.\n\nChange-Id: Idead9b4198c6b05d72bae60dee06e5aebc223822\nReference: https://github.com/PyCQA/bandit/pull/489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/1d7ca8a198419896346369311fccedea5a559f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/1d7ca8a198419896346369311fccedea5a559f66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
