)]}'
{"id":"openstack%2Foslo.policy~671113","triplet_id":"openstack%2Foslo.policy~master~Ife6600da240f830aa0080e3f214cedf1389ea512","project":"openstack/oslo.policy","branch":"master","topic":"test","hashtags":[],"change_id":"Ife6600da240f830aa0080e3f214cedf1389ea512","subject":"Add unit tests on cache handler","status":"MERGED","created":"2019-07-16 17:13:12.000000000","updated":"2021-04-16 12:12:31.000000000","submitted":"2021-04-16 12:12:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671113-test","meta_rev_id":"5bffaa18956a578918f68492aa0a0c36a6770640","_number":671113,"virtual_id_number":671113,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 12:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2021-02-09 14:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2021-04-16 10:33:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 12:25:33.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-12-02 15:33:22.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-08-18 18:31:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-14 16:08:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"3789d16171902d7de744dc226023a7da8b86e36e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-16 17:13:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4c447157ff908a9020c18a5061c7a603f20af11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 18:54:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/13/671113/1/check/tempest-full/2166eec/ : SUCCESS in 1h 38m 26s\n- tempest-full-py3 http://logs.openstack.org/13/671113/1/check/tempest-full-py3/2c860d9/ : SUCCESS in 1h 31m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/671113/1/check/openstack-tox-lower-constraints/4a4cfc1/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/13/671113/1/check/openstack-tox-pep8/857c286/ : FAILURE in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/13/671113/1/check/openstack-tox-py27/d12dbc1/ : FAILURE in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/13/671113/1/check/openstack-tox-py36/65730aa/ : SUCCESS in 3m 29s\n- openstack-tox-py37 http://logs.openstack.org/13/671113/1/check/openstack-tox-py37/2503960/ : SUCCESS in 4m 18s\n- openstack-tox-docs http://logs.openstack.org/13/671113/1/check/openstack-tox-docs/88717c6/html/ : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dd46a530e513e8225188a6b59626c0cd8c565571","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-16 21:14:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eecfb92c448702e7ea7532d1eb3c7b067d816197","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:00:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/13/671113/2/check/tempest-full/713ea74/ : SUCCESS in 1h 39m 52s\n- tempest-full-py3 http://logs.openstack.org/13/671113/2/check/tempest-full-py3/a22193b/ : SUCCESS in 1h 14m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/671113/2/check/openstack-tox-lower-constraints/9f93bad/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/13/671113/2/check/openstack-tox-pep8/33f064c/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/13/671113/2/check/openstack-tox-py27/d14a029/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/13/671113/2/check/openstack-tox-py36/4ba87c8/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/13/671113/2/check/openstack-tox-py37/3d2a97f/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/13/671113/2/check/openstack-tox-docs/612a917/html/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":2},{"id":"baacf3b4ed6a3a1735bbb38036b2becf063edf7a","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-07-22 12:25:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"41be048d7c51d66250705a67d823fd527f0511ec","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-02 15:33:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nA few minor things and questions inline.","accounts_in_message":[],"_revision_number":2},{"id":"1b4e9fc13171babe5d6a0c72990b97cdb99b9902","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-18 16:48:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f1464f54d7393c7df06111b7cc2cd6fd7b42480","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-18 17:08:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ca2683373011ffcf94d21214cecd31c92c222df7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 18:31:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/218f33bd70c54fb0af197e67172a32d0 : SUCCESS in 1h 17m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07d711b5d9a34067ab33a8873ea340ad : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/992ff8455709475a989c7afc4d68d77b : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d663aadb97b34e588df18edbb2d78aee : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b36aa1abef44453aed1a953f8a17d62 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a572d0928074845839f7e436a953fe5 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":4},{"id":"d0ff23ae695b15ccb18c255634c229f127447a85","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-14 16:08:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"78eb1c01777dcc947ed00c412cb51382c9adbfed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-09 14:32:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"81fb0ad7096cfd4828757b296835985672fedced","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-09 14:32:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7b76f19d16218c6aa8070249656f01725d9f7d8f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-09 14:33:02.000000000","message":"Patch Set 5: Code-Review+2\n\nAddressed my own nits plus some from bnemec previously","accounts_in_message":[],"_revision_number":5},{"id":"3f8dd04a4082a0058940ac94953b4ebbf2031ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:00:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/397ffadb4a9742a494036f57f8248554 : SUCCESS in 1h 16m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8beadbcc88024eefb39afcc732308bdd : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc7fddde5124c2cb322573fb1c5cf00 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7236c5c250e2420dbdddc20e8bb846f2 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57368c9acddb4d64b6099df129963b96 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0f5849f70ce47d8bb736a44feee94b7 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92adae57e2374f87a7410ba7cc6facfa : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":5},{"id":"f05415103c6a8066425948a43de6ee9f2c722cf0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:33:44.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a8b0b41e85c6f8d33c93d3f4da54275ab54e7a30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:33:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4a2383f6fa5e442862f24f4bcecc5c71fe3d6965","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:12:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9c7de7258b04e84a6f6dc69460f0c71 : SUCCESS in 1h 31m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/913b978b3f224599b8b67f5b52c0f6e8 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0681a4fb354d2185578d12f5d84c82 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53e1d634e5f24d5b871a6e1c627cf6ff : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd43576c50e34aa78bd883e44ad7381b : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e416e77003b4279b7992c1918e8682e : SUCCESS in 5m 49s\n- cross-nova-tox-py38 https://zuul.opendev.org/t/openstack/build/8741d8e4c00343948b863c088b98d7e0 : SUCCESS in 15m 12s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/c35d9862feee41b3831c3a778a36e453 : SUCCESS in 25m 08s\n- cross-neutron-tox-py38 https://zuul.opendev.org/t/openstack/build/0b42e72a997a453f91746e0be6a9a4e5 : SUCCESS in 31m 28s","accounts_in_message":[],"_revision_number":5},{"id":"5bffaa18956a578918f68492aa0a0c36a6770640","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:12:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4065a08ea401e0d95e8d40d9d735edf92edda861","revisions":{"e03775991097c826bad346c4c9e02e68bb580d7d":{"kind":"REWORK","_number":1,"created":"2019-07-16 17:13:12.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/13/671113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/13/671113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/1"}}},"commit":{"parents":[{"commit":"b7da7a92ad558947d6a133800f6049578e450b6a","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b7da7a92ad558947d6a133800f6049578e450b6a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-07-16 17:12:34.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-07-16 17:12:34.000000000","tz":120},"subject":"Add unit tests on cache handler","message":"Add unit tests on cache handler\n\nChange-Id: Ife6600da240f830aa0080e3f214cedf1389ea512\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/e03775991097c826bad346c4c9e02e68bb580d7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/e03775991097c826bad346c4c9e02e68bb580d7d"}]},"branch":"refs/heads/master"},"ef8973de80e9750a1d8a3953187a8e092b5b5377":{"kind":"REWORK","_number":2,"created":"2019-07-16 21:14:41.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/13/671113/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/13/671113/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/2"}}},"commit":{"parents":[{"commit":"b7da7a92ad558947d6a133800f6049578e450b6a","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b7da7a92ad558947d6a133800f6049578e450b6a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-07-16 17:12:34.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-07-16 21:14:32.000000000","tz":120},"subject":"Add unit tests on cache handler","message":"Add unit tests on cache handler\n\nChange-Id: Ife6600da240f830aa0080e3f214cedf1389ea512\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/ef8973de80e9750a1d8a3953187a8e092b5b5377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/ef8973de80e9750a1d8a3953187a8e092b5b5377"}]},"branch":"refs/heads/master"},"edc976470258c3631a50c396c72271a60a64caf4":{"kind":"REWORK","_number":3,"created":"2020-08-18 16:48:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/13/671113/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/13/671113/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/3"}}},"commit":{"parents":[{"commit":"a626ad12fe5a3abd49d70e3e5b95589d279ab578","subject":"Merge \"Clarify what exactly an \"access file\" is\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/a626ad12fe5a3abd49d70e3e5b95589d279ab578"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-07-16 17:12:34.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-18 16:48:33.000000000","tz":120},"subject":"Add unit tests on cache handler","message":"Add unit tests on cache handler\n\nChange-Id: Ife6600da240f830aa0080e3f214cedf1389ea512\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/edc976470258c3631a50c396c72271a60a64caf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/edc976470258c3631a50c396c72271a60a64caf4"}]},"branch":"refs/heads/master"},"22c4c4c6e7f99cd1c7a0c274ec4a4306f50c3c73":{"kind":"REWORK","_number":4,"created":"2020-08-18 17:08:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/13/671113/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/13/671113/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/4"}}},"commit":{"parents":[{"commit":"a626ad12fe5a3abd49d70e3e5b95589d279ab578","subject":"Merge \"Clarify what exactly an \"access file\" is\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/a626ad12fe5a3abd49d70e3e5b95589d279ab578"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-07-16 17:12:34.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-18 17:03:40.000000000","tz":120},"subject":"Add unit tests on cache handler","message":"Add unit tests on cache handler\n\nChange-Id: Ife6600da240f830aa0080e3f214cedf1389ea512\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/22c4c4c6e7f99cd1c7a0c274ec4a4306f50c3c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/22c4c4c6e7f99cd1c7a0c274ec4a4306f50c3c73"}]},"branch":"refs/heads/master"},"4065a08ea401e0d95e8d40d9d735edf92edda861":{"kind":"REWORK","_number":5,"created":"2021-02-09 14:32:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/671113/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/13/671113/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/13/671113/5"}}},"commit":{"parents":[{"commit":"de243e7a72097246a1c9be9072a4322df38927b2","subject":"Work on copy of registered rule instead of original object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/de243e7a72097246a1c9be9072a4322df38927b2"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-07-16 17:12:34.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-02-09 14:32:06.000000000","tz":0},"subject":"Add unit tests on cache handler","message":"Add unit tests on cache handler\n\nChange-Id: Ife6600da240f830aa0080e3f214cedf1389ea512\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/4065a08ea401e0d95e8d40d9d735edf92edda861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/4065a08ea401e0d95e8d40d9d735edf92edda861"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
