)]}'
{"id":"openstack%2Foslo.policy~694841","triplet_id":"openstack%2Foslo.policy~stable%2Ftrain~Ifd14993bc84e83c13abab3456fbf670c06e5806f","project":"openstack/oslo.policy","branch":"stable/train","topic":"bug/1843931-stable/train","hashtags":[],"change_id":"Ifd14993bc84e83c13abab3456fbf670c06e5806f","subject":"Fix reference cycle caused by deprecated sample override","status":"MERGED","created":"2019-11-18 20:23:22.000000000","updated":"2019-12-12 22:52:40.000000000","submitted":"2019-12-12 22:50:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694841-1576191055075-8b70566f","meta_rev_id":"a14f0f21b23ccbd9eb3cf8c49abb0a208b4661ad","_number":694841,"virtual_id_number":694841,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 22:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":2,"date":"2019-12-12 20:59:04.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-11-26 10:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-12-12 20:59:04.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 10:27:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-12 20:59:04.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-12-12 22:50:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0afeec9735876210ff690f441f55357356cc271b","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-11-18 20:23:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ecefdbc35e7995f76db88774f8840d88a859b04c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 22:03:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/88dd7e2bbc38400cbab0c91a7243b14f : SUCCESS in 1h 34m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0dfb96a35a942279881cedf47e9c758 : SUCCESS in 1h 38m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df737ebc8fed4107baa20e8795ccddb2 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21628b176e2a4f1f942e3b1c03b32854 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/701dca8ea83041caa21eed66af5fadf4 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e77b6f0bde244fe988fe2de231ced097 : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40ec913f9f124718b370062284a888ac : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472a1c8794174d3fabc67148e1c6b241 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"23d91ba8b076b1e526aa1da7e20337669c4d6faa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-26 10:27:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7c5d5171c8f41156a64f25f08ea3057fbfa220c","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-12 20:59:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis has been merged in Stein for a while, so it should definitely be in Train.","accounts_in_message":[],"_revision_number":1},{"id":"dc6eda178542ee09b968e2a002ac85f209c4f029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 20:59:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5973f1026611961509dce60706614f7cbe765461","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:50:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9b8fe8228cbb4b089cf0624a0f14b624 : SUCCESS in 1h 45m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7622132ce1e419a8af5dc383850de74 : SUCCESS in 1h 35m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad2b60a241f24b3a958260455508e444 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ebfade41e964a2eaefcf4976dc39a4e : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01230009f162453f933c968cd1810167 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dbe133ce1bc427e873df63deca7bb77 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b0a4279069f47a9ad6b099ef674967f : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81a8f6639d34650ba0592429bd07231 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":1},{"id":"829232402efb65ac11164e60d90e066fb8564829","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:50:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"618cc3f3557133b889cfc74b29a094403653a041","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:52:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dce0ffbf5c75424699b959871b82c1e5 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3898da6d8c54229aa60c0e3571c7ec0f847ca8ea","revisions":{"3898da6d8c54229aa60c0e3571c7ec0f847ca8ea":{"kind":"REWORK","_number":1,"created":"2019-11-18 20:23:22.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/41/694841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/41/694841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/41/694841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/41/694841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/41/694841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/41/694841/1"}}},"commit":{"parents":[{"commit":"b3ab3fb4b5d91a3b8b8387f4f37ce407fe18e92f","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b3ab3fb4b5d91a3b8b8387f4f37ce407fe18e92f"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-09-13 20:04:56.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"openstack@nemebean.com","date":"2019-11-18 20:23:22.000000000","tz":0},"subject":"Fix reference cycle caused by deprecated sample override","message":"Fix reference cycle caused by deprecated sample override\n\nIn the sample policy generator, we create a rule that maps the\ndeprecated name of a policy to the new rule name. For example:\n\nidentity:old_rule: rule:identity:new_rule\n\nHowever, in the policy code, if we see an override of a deprecated\nname and no override for the new name, we apply the value of the\ndeprecated name to the new name. In the above case, this results\nin us creating a rule that looks like:\n\nidentity:new_rule: rule:identity:new_rule\n\nwhich is a circular reference and nonsense.\n\nTo fix this, I added a check to the deprecated rule logic that looks\nfor instances where the old override is just a reference to the new\nrule. If that\u0027s the case, then we don\u0027t need to do anything because\nit\u0027s already doing the right thing.\n\nChange-Id: Ifd14993bc84e83c13abab3456fbf670c06e5806f\nCloses-Bug: 1843931\n(cherry picked from commit 82a2c8d8b71896b1f8b7f33c560681367ae76755)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/3898da6d8c54229aa60c0e3571c7ec0f847ca8ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/3898da6d8c54229aa60c0e3571c7ec0f847ca8ea"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
