)]}'
{"id":"openstack%2Foslo.policy~734151","triplet_id":"openstack%2Foslo.policy~stable%2Fussuri~I8a6f8e971d881365c41ea409966723319d5b239a","project":"openstack/oslo.policy","branch":"stable/ussuri","topic":"bug/1880959","hashtags":[],"change_id":"I8a6f8e971d881365c41ea409966723319d5b239a","subject":"Reload files in policy_dirs on primary file change","status":"MERGED","created":"2020-06-08 16:51:17.000000000","updated":"2020-06-22 17:37:55.000000000","submitted":"2020-06-22 17:35:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734151-1592847316333-3f73efbb","meta_rev_id":"5d499da31b508b543fe7121680d1b87a2d7e96ad","_number":734151,"virtual_id_number":734151,"owner":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-22 15:57:10.000000000","_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 17:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-22 15:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":2,"date":"2020-06-15 10:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-22 15:55:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 10:16:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-22 15:55:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-22 17:35:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46d856760945e8ce1c7dc141b7d2644a71db656e","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2020-06-08 16:51:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c0fdfc410c835da73b950f216aaf3502450bf7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:03:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2474f8824d443bdaaf6eeb703946732 : SUCCESS in 1h 10m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7736f22bcca4fa4bbed7ad0e9c0fd8e : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f13c79d5eff74065a11826d777e2cf94 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/187b4904bf6148419e72fb0b38b75d01 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdd5f1c082da425ca335b0c708c59eae : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2298c5f8aa444a84a7177a2b03019549 : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6a14ae668924059b2656a08a1df187e : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5a92aae5c726cd7ad4751ebf6bd162a9e4c52e7e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-15 10:16:19.000000000","message":"Patch Set 1: Code-Review+2\n\nclean backport","accounts_in_message":[],"_revision_number":1},{"id":"7381fd91b1b5a43410754e5928ccc5762bf3f85e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 15:55:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a0ec6645d8add29f2c3c14a81a1d9b66657dbd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 15:55:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1235d2c4d02aaef26238bb420c18430273af81d8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 15:55:45.000000000","message":"Patch Set 1:\n\nFor future reference, please do not cherry-pick to older branches until it has merged to the more recent branch first.","accounts_in_message":[],"_revision_number":1},{"id":"ad7daa6f7498fd12797952398151717dd7507372","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2020-06-22 15:57:10.000000000","message":"Patch Set 1:\n\nUnderstood, thank you!","accounts_in_message":[],"_revision_number":1},{"id":"c1cc90d8031afb41a22d652af81f8501bf8408a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 17:35:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff7028a54ecc431c8951eeb2885a4e1d : SUCCESS in 1h 35m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c756c581a3543ecb327774f5cffe5b8 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bdce46a65a9452d8296649d6d26bd24 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ab146a33ffa4e23b204c0b0dc60191e : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29ecd735b87043fc93b3a52df3bdc582 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3564d6bd31a46478563f55e55cafbc5 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87f4b5f708dc4cb48badcc226adbd421 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"76e332c7fe5e8e33e8cc1677cf7cac0c09052367","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 17:35:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f7c0cec822fb2eb60783d55ea450394da441b63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 17:37:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e46f1b4e33549358a1c0afd3b835c25 : SUCCESS in 1m 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72274a12494a4b5c8be9e433710a4452 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8c138e69d9189345a7366a10b0779ef70b7250e","revisions":{"c8c138e69d9189345a7366a10b0779ef70b7250e":{"kind":"REWORK","_number":1,"created":"2020-06-08 16:51:17.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/51/734151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/51/734151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/51/734151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/51/734151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/51/734151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/51/734151/1"}}},"commit":{"parents":[{"commit":"d4a2c64fad8b0d8d2df6a447c6c9826deaf24593","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/d4a2c64fad8b0d8d2df6a447c6c9826deaf24593"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2020-05-27 14:06:25.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2020-06-08 16:50:09.000000000","tz":180},"subject":"Reload files in policy_dirs on primary file change","message":"Reload files in policy_dirs on primary file change\n\nIt was determined that rules from policy files located in the directory\nspecified in the policy_dirs option (/etc/\u003cconfig_dir\u003e/policy.d by\ndefault) are not re-applied after the rules from the primary policy file\nis re-applied due to a change.\n\nThis change introduces additional behavior to make sure the rules from\npolicy_dirs are reapplied if there is a change to the primary policy\nfile.\n\nChange-Id: I8a6f8e971d881365c41ea409966723319d5b239a\nCloses-Bug: #1880959\nRelated-Bug: #1880847\n(cherry picked from commit 75677a31108243e0adddc89f1fbf669053f9573b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/c8c138e69d9189345a7366a10b0779ef70b7250e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/c8c138e69d9189345a7366a10b0779ef70b7250e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
