)]}'
{"id":"openstack%2Foslo.policy~760049","triplet_id":"openstack%2Foslo.policy~master~I138e5cf6deb82ad29a2bf70392c1859354b84800","project":"openstack/oslo.policy","branch":"master","topic":"bug_fix","hashtags":[],"change_id":"I138e5cf6deb82ad29a2bf70392c1859354b84800","subject":"Fix grammar issues","status":"MERGED","created":"2020-10-28 05:21:27.000000000","updated":"2020-11-17 18:23:08.000000000","submitted":"2020-11-17 18:21:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760049-1605637301330-bced994f","meta_rev_id":"7f32ad7dc3a6fc174563a389c5c8630f1fd0d9cb","_number":760049,"virtual_id_number":760049,"owner":{"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-17 18:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-11-02 14:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2020-11-17 18:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-12 17:53:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2020-11-17 18:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},{"value":1,"date":"2020-11-02 16:04:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 19:05:46.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2020-11-02 14:58:05.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-11-02 16:04:42.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2020-11-12 17:53:16.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-11-17 18:05:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-17 18:21:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56e697a417a6b9fe8af4f72a5d9db22b7e597668","author":{"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},"date":"2020-10-28 05:21:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00abf8f62885fca8b2211a24d39a2c372e59384f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 05:28:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22004c9433c549d189b05189798a2f50 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c84dd979d04346d487aeeaff0340940b : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"109e59c9ec9404ea094dbfe9f399477bc1d35b3c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-28 10:36:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nLGTM but it could be worth to grab feedback from other core that are native speakers.","accounts_in_message":[],"_revision_number":1},{"id":"dfbf25d9980113711c6303b17a9165f3476980bf","author":{"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},"date":"2020-10-28 21:46:11.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThank you for reviewing the changes. I added the reasons for the changes. I am also adding a core member for the review.","accounts_in_message":[],"_revision_number":1},{"id":"0acf8d6e073f2468cde00d54452f153781601764","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2020-10-30 19:05:46.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e13e48fa5540310b8167c3d6bfd388ef3fc6b478","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-11-01 13:17:10.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nMost of this is definitely an improvement, thank you!\n\nI have one suggestion, one question, and one nit inline.","accounts_in_message":[],"_revision_number":1},{"id":"bb7785a21e394822b1a2f9e443a6893560384955","author":{"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},"date":"2020-11-02 02:01:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53942be01e840e25da953a669607d3fed8bc65bf","author":{"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},"date":"2020-11-02 02:08:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"648fd5f877f123e5fa975f3dc3f8110e5579ca90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 02:18:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/501f660ba50b4e68b5fff44625736a62 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852d790fb0ef43479440e34b88bc3b81 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"5a5ca5c683de306c51a1e7a8128882d9bddbe554","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-11-02 14:58:05.000000000","message":"Patch Set 3: Code-Review+2\n\nlgtm. I think the previous comments have been addressed and this cleans up a number of issues.","accounts_in_message":[],"_revision_number":3},{"id":"9041adbd6972fe99b9e1885111bc9d49f28eb1e3","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-11-02 16:04:42.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"7daa55cf971a36318d731e792e7e2728b6cd754e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-12 17:53:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c7faee88f2c1f465c1e6358d9e0f943cfa223aab","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-17 18:05:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"13ddece82c2832c31ff75440bb06d70032284231","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-17 18:05:12.000000000","message":"Patch Set 3:\n\nThank you","accounts_in_message":[],"_revision_number":3},{"id":"55d22c1280b110cccfdda5a478b81cd957f890df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 18:05:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1c9a26088b5f71b827d70686ddcc9b3fe4cecf31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 18:21:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b24ef94891ce43a297bee3d26f64601e : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f7209b7651f4599b1c0c626c109ba16 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"531a6854cc8d72b944d4824e9d3cb37462e6daf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 18:21:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e3f5d5d552b41dfd5e7edbc69dab82b770a73971","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 18:23:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6fb94d4754409f8b0f08e0447b0e02 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"96cfafb261acba2524461c481b68c98831697cb9","revisions":{"05fa3f112ba8398cc9b1b121243e16c088b060a1":{"kind":"REWORK","_number":1,"created":"2020-10-28 05:21:27.000000000","uploader":{"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},"ref":"refs/changes/49/760049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/49/760049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/1"}}},"commit":{"parents":[{"commit":"266ee36d3387247f8cbc57327a71f86a181fd45f","subject":"Adding pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/266ee36d3387247f8cbc57327a71f86a181fd45f"}]}],"author":{"name":"Qi Zhang","email":"qizhangapp@gmail.com","date":"2020-10-28 05:15:02.000000000","tz":-420},"committer":{"name":"Qi Zhang","email":"qizhangapp@gmail.com","date":"2020-10-28 05:15:02.000000000","tz":-420},"subject":"Fix grammar issues","message":"Fix grammar issues\n\nChange-Id: I138e5cf6deb82ad29a2bf70392c1859354b84800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/05fa3f112ba8398cc9b1b121243e16c088b060a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/05fa3f112ba8398cc9b1b121243e16c088b060a1"}]},"branch":"refs/heads/master"},"27f65e5a668c096aaef7614c959e06b38c40a1e9":{"kind":"REWORK","_number":2,"created":"2020-11-02 02:01:42.000000000","uploader":{"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},"ref":"refs/changes/49/760049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/49/760049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/2"}}},"commit":{"parents":[{"commit":"266ee36d3387247f8cbc57327a71f86a181fd45f","subject":"Adding pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/266ee36d3387247f8cbc57327a71f86a181fd45f"}]}],"author":{"name":"Qi Zhang","email":"qizhangapp@gmail.com","date":"2020-10-28 05:15:02.000000000","tz":-420},"committer":{"name":"Qi Zhang","email":"qizhangapp@gmail.com","date":"2020-11-02 02:00:38.000000000","tz":-480},"subject":"Fix grammar issues","message":"Fix grammar issues\n\nChange-Id: I138e5cf6deb82ad29a2bf70392c1859354b84800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/27f65e5a668c096aaef7614c959e06b38c40a1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/27f65e5a668c096aaef7614c959e06b38c40a1e9"}]},"branch":"refs/heads/master"},"96cfafb261acba2524461c481b68c98831697cb9":{"kind":"REWORK","_number":3,"created":"2020-11-02 02:08:49.000000000","uploader":{"_account_id":32519,"name":"Qi Zhang","email":"qizhangapp@gmail.com","username":"qizhangapp"},"ref":"refs/changes/49/760049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/49/760049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/49/760049/3"}}},"commit":{"parents":[{"commit":"266ee36d3387247f8cbc57327a71f86a181fd45f","subject":"Adding pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/266ee36d3387247f8cbc57327a71f86a181fd45f"}]}],"author":{"name":"Qi Zhang","email":"qizhangapp@gmail.com","date":"2020-10-28 05:15:02.000000000","tz":-420},"committer":{"name":"Qi Zhang","email":"qizhangapp@gmail.com","date":"2020-11-02 02:08:33.000000000","tz":-480},"subject":"Fix grammar issues","message":"Fix grammar issues\n\nChange-Id: I138e5cf6deb82ad29a2bf70392c1859354b84800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/96cfafb261acba2524461c481b68c98831697cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/96cfafb261acba2524461c481b68c98831697cb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
