)]}'
{"id":"openstack%2Foslo.policy~761349","triplet_id":"openstack%2Foslo.policy~master~Ibd2503db55739a9b8a3011534a7b51db16282001","project":"openstack/oslo.policy","branch":"master","topic":"tox_constraints_file","hashtags":[],"change_id":"Ibd2503db55739a9b8a3011534a7b51db16282001","subject":"Use TOX_CONSTRAINTS_FILE","status":"ABANDONED","created":"2020-11-04 09:49:47.000000000","updated":"2021-02-08 17:31:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d8e46e089fd1fd2ac203a1c3fe98aaab273a54de","_number":761349,"virtual_id_number":761349,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-14 11:30:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2021-02-05 07:28:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-12-11 15:38:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-12-11 15:38:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-04 11:49:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 15:38:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-02-05 07:28:39.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"f0c1cd4ed4c824916f2cbc098fecb7a2675b1110","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-04 09:49:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dde170737ecd42d46151b43ecf0cd7047269b565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 11:49:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f88de406e334b74ae41de80bbbd67af : SUCCESS in 1h 30m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/294bcf32d7aa48ca89dc49aa78058a3a : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09393b27cee4ba088f14784622f436c : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f8d6962821e4b9ca201d3d9c90ad676 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af41a8d720624cacb6f9b6d8be07a7b9 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28dc337a4f99414c9da216adf45e2aee : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b077d812804f908b70325c17c07a3b : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e6b303d4aa14d0ca8e7e2e3cb4c2815 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"68cd5cb13a69d719061e2b7aed1671255e1a8ba8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 15:38:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84d85a27e40039b1a5589d58733b411b27c993b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 15:39:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14a0efed6cc2773bbbf781bb1abfbef9ec6596e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 17:32:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5fbc84201524251ac081863230ebbe4 : SUCCESS in 1h 28m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ffecf7c105b480c94a43d6765d57706 : FAILURE in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c8c3da5c3a9435aa728276a3cdddb27 : FAILURE in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba41c77c879c43f89a1f43bc16fd50f6 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21f9db247e4c4ce5a276ba32e37eb65f : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d054590fccef411692e18ae00b919986 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1d9714504f14310a83c9b10036e5b7d : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"798f6577850058c4119b3ad3c0cb49fdaeef8692","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-14 09:34:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03de3a642ad0d01f38fd4a88e075be079d85666f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 11:30:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86402b7d21cf474a958111f7f9c734b5 : SUCCESS in 1h 13m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/befe5b3c4137467aacc71a1a568de5ac : FAILURE in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c04485a32214341b3a10cb699e6ef05 : FAILURE in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46a04d57bbab41c5914282df101b3489 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94b4ac59c305449e8f932d5c6744f5a3 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5db8b639a6674e98b01cc43c39bde3f2 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc27c5555f334214852f3f341bc45b5b : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6ae51b2bd8f491b8f90bc1d9a10c1ed : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c96be3b186c5a1e577631ce6335621a66c81479d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-20 11:53:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e6919bc05c55d21cf707cc15cc2f8bb2d0e1f7df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 13:24:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54ac455457554fa48db66be0018a647c : SUCCESS in 1h 21m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bb6676746ae4895a848368b81bd0073 : FAILURE in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e25a22e5fdd944fea0350389d4ee6e48 : FAILURE in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04cee171bfbe453bbfdb6f30bc36a274 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c82d084ffc147ee89f45d7d713715f7 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ba353473bf44b99aa2d65e2392f9106 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa753642d4c43469b559c6984f0efb1 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/847c07e625ea47e6b63f23fcf255d8e2 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"94b70053e875f542f153aea5f91b1e7e66526376","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 16:20:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f2c606bd4217e61bcfb9e763973145505caab62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 18:04:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2a22915e1c142ce9bdb6f691a5dcf78 : SUCCESS in 1h 32m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1138dad436a4327af121cb6ec49d4a7 : FAILURE in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd67565815a479c88b90fe8c0ad58b6 : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/164c3dc5ac5646e1bbb4af68b1048dcd : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcc817ec5b2e477ea65d115c3d075217 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62282b56ba9e430d944cfc02d1e9d01d : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8213694d1a9944b48fb7124afd2fb088 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39d01d5eb39f4369a3c8de6d8076f6d8 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cc684840bd7fac4732957c5a5705038f61eb6677","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 08:10:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7928307023e5a0ed982e4bd3bbb51a6a6587823f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 09:45:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a05fa7ef1ed84d9d9349b47d3bd76a31 : SUCCESS in 1h 28m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/015b157f482644a4b9ae553d8ae9e79a : FAILURE in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3afcccbe19445ff9b89281bfba55c38 : FAILURE in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd68e61fd1534ebf9ac4b3a1ba35bffc : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/602a491ba5f34564b250d0131917d0d9 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb249494b61845e880abf7bfabe8d679 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a638aa36ebf4c40b5877d87d18e7967 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd7b65c737c84755b0ecc9c3e4efe4cb : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"03cfc76a16d4fa2b173762fac152a52c9987753c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-02-05 07:28:39.000000000","message":"Patch Set 1: Code-Review-1\n\nThis patch is not needed anymore as it is fixed by\nhttps://review.opendev.org/c/openstack/oslo.policy/+/760709 .\nCould you abandon this patch?","accounts_in_message":[],"_revision_number":1},{"id":"d8e46e089fd1fd2ac203a1c3fe98aaab273a54de","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-08 17:31:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0529d84c22acfea957a80cdba2abb5adf0fca2ca","revisions":{"0529d84c22acfea957a80cdba2abb5adf0fca2ca":{"kind":"REWORK","_number":1,"created":"2020-11-04 09:49:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/49/761349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/49/761349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/761349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/761349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/49/761349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/49/761349/1"}}},"commit":{"parents":[{"commit":"30e788c592067f8896e267399e734c3dd8851f52","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/30e788c592067f8896e267399e734c3dd8851f52"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-04 09:49:41.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-04 09:49:41.000000000","tz":60},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\n[1] https://review.opendev.org/#/c/722814/\n[2] https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nChange-Id: Ibd2503db55739a9b8a3011534a7b51db16282001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/0529d84c22acfea957a80cdba2abb5adf0fca2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/0529d84c22acfea957a80cdba2abb5adf0fca2ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
