)]}'
{"id":"openstack%2Foslo.policy~763261","triplet_id":"openstack%2Foslo.policy~master~If72b2fcc3cfd8116b575ed7b9e3870df634fd9af","project":"openstack/oslo.policy","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"If72b2fcc3cfd8116b575ed7b9e3870df634fd9af","subject":"Add policy file selection logic when default changing to yaml","status":"MERGED","created":"2020-11-18 20:45:56.000000000","updated":"2020-11-23 02:38:53.000000000","submitted":"2020-11-23 02:37:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763261-policy-json-to-yaml","meta_rev_id":"161c3ba1d24911ccadf4c25b42e4a694ed0c4f2d","_number":763261,"virtual_id_number":763261,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-23 02:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"all":[{"value":2,"date":"2020-11-23 01:20:13.000000000","_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":2,"date":"2020-11-20 13:06:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"all":[{"value":1,"date":"2020-11-23 01:20:13.000000000","_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-11-23 01:20:13.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 19:39:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 13:06:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-20 14:50:23.000000000","updated_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"reviewer":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"state":"REVIEWER"}],"messages":[{"id":"23958df1e627c05da20d8841086c85c842bae022","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-18 20:45:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77f31f2b3ff4157a91fa013a85b7388a585262b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 22:19:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0b187fc7f4c41f4afcd738b46811ae0 : SUCCESS in 1h 31m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffce5d5c01574e3290d8efcd2f47536d : FAILURE in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007d2d5bef40494da3425f3fdd614d6a : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07cee717d8994efa9b906f1619cf70f6 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8688e62aee144b22a2d8b0a2ed261880 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12a7a5d2254f46cebb6e0a568c4cd694 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e9193e7c0234d3c88c12c26867e9504 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d2ab984c77f626db09b163fec10750245ecefc15","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-18 23:50:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"655bee3462ef046c29f8742fda3d80b3c3936bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 01:18:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/60fd609ca9214647869344f290430bf5 : FAILURE in 2m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5006d154caad41199e7c28d6678055ef : SUCCESS in 1h 24m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e900ffdcc26940d183b74f657c72fb4c : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04484e9c12ea4e5fa9e8068c97d2e574 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16264d9b797e45e0904322892d925c1e : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e04ce3ef96b64a82a47d51c0fd7441ba : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ef331b76f38478c968c4458901c8925 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5602de0dc294fabac920f103c0db816 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"7a09db8529385cf14dd912c7e89adb0d2500f90f","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2020-11-19 16:20:43.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI think you can drop the constraint check.","accounts_in_message":[],"_revision_number":2},{"id":"18aa26c928652394c076f3d39afbfc0d17325e74","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-19 16:26:13.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)\n\nneed to update requirement.txt too","accounts_in_message":[],"_revision_number":2},{"id":"528aa34d3c4150952b2320f0469c116acfe76ae3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-19 17:30:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3588e6f96d1f60e804ab83e5c0760749716baa65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 19:39:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9f909665e4ab46bfb4292d369ed0e2b7 : SUCCESS in 4m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed3862af43f54237912eb9b9419dd1d5 : SUCCESS in 1h 50m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd303b299c1b4b86823447af766290f3 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/831a5b71e670434bb5b0efcc23376155 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/140829b5081043e1b255c107714e0181 : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec4ce0b777ca4b9fa40839c2f1c8e208 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86d4629a7095458982113e33576280a6 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40095d1fa36343a5a84deaace7100822 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":3},{"id":"28dca37f4e9e8cfc95245e038436f48955bf2486","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-20 13:06:25.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":3},{"id":"7920304ab3d841328eb5f60e06fd75b6655d96de","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2020-11-20 14:50:23.000000000","message":"Patch Set 3: Backport-Candidate-1\n\n(1 comment)\n\nI\u0027ll remove the -1 if you can address my concerns in comments.","accounts_in_message":[],"_revision_number":3},{"id":"7395874419adda3680b766cdfea882708d2c464e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-22 23:53:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7acf3f10366a4baeae00c0826a081694beb3c24","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2020-11-23 01:20:13.000000000","message":"Patch Set 3: -Backport-Candidate Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e03fef6018b3b3908dcf1ee0bcd5a712976fc74d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 01:20:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ec16d5af56ce8e84b6b52c4b7491821c63de927a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 02:37:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7c5ba6a11c264d3089252863e0406766 : SUCCESS in 1m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26d1ab918d9e4a28b1d4c48bca56a6d6 : SUCCESS in 1h 16m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58c6ffa629f44ae5b5bad9ecee79bfca : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d16141096146adb258adf21acdbd4f : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4901dec6c8d646e58d870bb03651e46e : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2692cdc54b824e2fb64f82b77eae3db7 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2690e2c84faa4c9b90edc57285536070 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3},{"id":"fd549cc0e0bc1b13e079505d5a4e638f79c4429a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 02:37:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"161c3ba1d24911ccadf4c25b42e4a694ed0c4f2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 02:38:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5981ee10a3224068850c9ee76c40dc44 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5180e9674f30a44a530deeabbc02804bec66351d","revisions":{"c8401d73f81fec75dfb7c04239d441bccf816556":{"kind":"REWORK","_number":1,"created":"2020-11-18 20:45:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/61/763261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/61/763261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/1"}}},"commit":{"parents":[{"commit":"5917863ced38c2f6c92d7767801e80e67bfbee03","subject":"Merge \"Fix grammar issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/5917863ced38c2f6c92d7767801e80e67bfbee03"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-18 20:37:47.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-18 20:37:47.000000000","tz":-360},"subject":"Add policy file selection logic when default changing to yaml","message":"Add policy file selection logic when default changing to yaml\n\nAs part of community goal[1], each services are changing the default\nvalue of \u0027CONF.oslo_policy.policy_file\u0027 config option from \u0027policy.json\u0027\nto \u0027policy.yaml\u0027. oslo policy select the default value from\nCONF.oslo_policy.policy_file which will be policy.yaml as service will\nstart changing the default. To avoid breaking the existing deployment which\nare relying on old default (policy.json) file, a new fallback logic\nis implemented. If new default file \u0027policy.yaml\u0027 does not exist but old\ndefault \u0027policy.json\u0027 exist then fallback to use old default file.\n\nEach services are going to add upgrade checks and warnings for using JSON\nformatted policy file so in future we cna remove this fallback logic.\n\nThis logic was done in nova in Victoria cycle when nova changed the\ndefault value - https://review.opendev.org/#/c/748059/ . Moving this\nto oslo policy side will avoid the duplication on services side.\n\nAlso it provides a flag to disable this fallback.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: If72b2fcc3cfd8116b575ed7b9e3870df634fd9af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/c8401d73f81fec75dfb7c04239d441bccf816556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/c8401d73f81fec75dfb7c04239d441bccf816556"}]},"branch":"refs/heads/master"},"d69c8421a7949e428d207c51896c12a06d735ce7":{"kind":"REWORK","_number":2,"created":"2020-11-18 23:50:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/61/763261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/61/763261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/2"}}},"commit":{"parents":[{"commit":"5917863ced38c2f6c92d7767801e80e67bfbee03","subject":"Merge \"Fix grammar issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/5917863ced38c2f6c92d7767801e80e67bfbee03"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-18 20:37:47.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-18 23:50:48.000000000","tz":-360},"subject":"Add policy file selection logic when default changing to yaml","message":"Add policy file selection logic when default changing to yaml\n\nAs part of community goal[1], each services are changing the default\nvalue of \u0027CONF.oslo_policy.policy_file\u0027 config option from \u0027policy.json\u0027\nto \u0027policy.yaml\u0027. oslo policy select the default value from\nCONF.oslo_policy.policy_file which will be policy.yaml as service will\nstart changing the default. To avoid breaking the existing deployment which\nare relying on old default (policy.json) file, a new fallback logic\nis implemented. If new default file \u0027policy.yaml\u0027 does not exist but old\ndefault \u0027policy.json\u0027 exist then fallback to use old default file.\n\nEach services are going to add upgrade checks and warnings for using JSON\nformatted policy file so in future we cna remove this fallback logic.\n\nThis logic was done in nova in Victoria cycle when nova changed the\ndefault value - https://review.opendev.org/#/c/748059/ . Moving this\nto oslo policy side will avoid the duplication on services side.\n\nAlso it provides a flag to disable this fallback.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: If72b2fcc3cfd8116b575ed7b9e3870df634fd9af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/d69c8421a7949e428d207c51896c12a06d735ce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/d69c8421a7949e428d207c51896c12a06d735ce7"}]},"branch":"refs/heads/master"},"5180e9674f30a44a530deeabbc02804bec66351d":{"kind":"REWORK","_number":3,"created":"2020-11-19 17:30:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/61/763261/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/61/763261/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/61/763261/3"}}},"commit":{"parents":[{"commit":"5917863ced38c2f6c92d7767801e80e67bfbee03","subject":"Merge \"Fix grammar issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/5917863ced38c2f6c92d7767801e80e67bfbee03"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-18 20:37:47.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-19 17:30:23.000000000","tz":-360},"subject":"Add policy file selection logic when default changing to yaml","message":"Add policy file selection logic when default changing to yaml\n\nAs part of community goal[1], each services are changing the default\nvalue of \u0027CONF.oslo_policy.policy_file\u0027 config option from \u0027policy.json\u0027\nto \u0027policy.yaml\u0027. oslo policy select the default value from\nCONF.oslo_policy.policy_file which will be policy.yaml as service will\nstart changing the default. To avoid breaking the existing deployment which\nare relying on old default (policy.json) file, a new fallback logic\nis implemented. If new default file \u0027policy.yaml\u0027 does not exist but old\ndefault \u0027policy.json\u0027 exist then fallback to use old default file.\n\nEach services are going to add upgrade checks and warnings for using JSON\nformatted policy file so in future we cna remove this fallback logic.\n\nThis logic was done in nova in Victoria cycle when nova changed the\ndefault value - https://review.opendev.org/#/c/748059/ . Moving this\nto oslo policy side will avoid the duplication on services side.\n\nAlso it provides a flag to disable this fallback.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: If72b2fcc3cfd8116b575ed7b9e3870df634fd9af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/5180e9674f30a44a530deeabbc02804bec66351d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/5180e9674f30a44a530deeabbc02804bec66351d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
