)]}'
{"id":"openstack%2Foslo.policy~799539","triplet_id":"openstack%2Foslo.policy~master~I9f35db891d9a0812a4aa0b538505f6ef65ecd0d2","project":"openstack/oslo.policy","branch":"master","attention_set":{},"removed_from_attention_set":{"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2024-07-18 20:00:34.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-07-18 20:00:34.000000000","reason":"Change was abandoned"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2022-04-27 13:31:29.000000000","reason":"Daniel Bengtsson replied on the change"}},"hashtags":[],"change_id":"I9f35db891d9a0812a4aa0b538505f6ef65ecd0d2","subject":"Make policy rule default change warning configurable","status":"ABANDONED","created":"2021-07-05 19:16:00.000000000","updated":"2024-07-18 20:00:34.000000000","total_comment_count":13,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"1886fa0371d9db0a446cdcf5ecf35a0c7aa6afd4","_number":799539,"virtual_id_number":799539,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-25 14:19:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":-1,"date":"2022-04-25 17:59:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 20:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 10:14:03.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-07-06 10:30:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2021-07-08 15:13:44.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"CC"},{"updated":"2021-08-02 15:35:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-02 16:27:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2021-08-18 22:23:26.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2022-04-25 13:52:25.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"}],"messages":[{"id":"adf5050e7f414d33ba3422da455478219dfc7ab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-05 19:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"092d9010dab969a67a518036ccb71d41ef3a8888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 20:32:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/083ad659c67c42caad2184bb714312b7 : SUCCESS in 1h 15m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81752fab805b424aab9666e8dd73fa35 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db3f50dc17b645eeb92824c29978ced5 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7328a969f03243bc80d7d24dc22fa94a : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6cd7478b5644bc5a01304d0e261b4aa : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac2ce6cc2594f6399f7343e43f633a6 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffce3565a6834af69d6fa8aadea1f43d : SUCCESS in 4m 02s\n- cross-nova-tox-py38 https://zuul.opendev.org/t/openstack/build/47ea7eb2f7474419a93d837b1ea4f476 : SUCCESS in 12m 46s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/447963d70248408199aa324da9a3d081 : SUCCESS in 22m 12s\n- cross-neutron-tox-py38 https://zuul.opendev.org/t/openstack/build/e820fa220957417aa68766069a0985d9 : SUCCESS in 22m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cf8129b55ee72aa60412c76ccdce40ea3b162a34","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-06 10:14:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nSome thoughts inline.","accounts_in_message":[],"_revision_number":1},{"id":"d2e42e6eefef2f676e1e60b853a4e7b940bf4159","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-06 16:02:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"548db136b224db66e81bbb9584efb0f75cbf7391","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-07-08 15:13:44.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nA bit of historical context around this inline.","accounts_in_message":[],"_revision_number":1},{"id":"4f940ffd988c71d121bcab0eb29c2d4c298e7e54","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 17:36:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0a3aa41c6894ede172e8ac717f8f152b466c757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 17:42:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"526dbf7d05605e7fa0d29a1f646cf464da55a34f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 18:48:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea41049c405e4b1dbb3c19f022d488a0 : SUCCESS in 1h 04m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9278d8535374beaa54f5d7c793b45aa : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4de1e2899ad499f8c9c8dab8f3f37d2 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a936bea2965b46d293f9f5d9e3f7a0d7 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66716486e4cc448c872550458bd7d6d7 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74bf3880789b43b0aa0998f2efebd0ba : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05815b990db142228d63668eae83e2b1 : SUCCESS in 4m 17s\n- cross-nova-tox-py38 https://zuul.opendev.org/t/openstack/build/7ac8bcaf48644ff2badd8f12e202845f : SUCCESS in 13m 22s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/69e0575e66c84ee2832b25ba0aae5ec6 : SUCCESS in 19m 07s\n- cross-neutron-tox-py38 https://zuul.opendev.org/t/openstack/build/4611582d1df94811988cb1238e428ac7 : SUCCESS in 23m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b71fcaf3cbd0e75356509d7f6c88535f3b3699f4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-02 15:35:33.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"f54a9e213366f04012c4201058043719d463d5bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-02 16:27:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce1d25330ef209be8fa4d3dabd2359eda4ef2332","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-08-02 18:07:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79f207f8e5dbf04de51012ac4366fa369f2036da","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-19 16:56:35.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"89f8b088bb2d7c13660df5abe0fd63c9555a1430","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-25 13:52:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5348c5358d642d20acca1611779da7dad1899d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-25 13:57:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"57b2b89693200e05be3f784c1400c95b26057f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:19:10.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2488eb2e63a485b8a405fc45b58a1ed : FAILURE in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f134e37bb2c44de78f56d83f5cf7002a : FAILURE in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c4375f9694b4dcca704df8852592c69 : FAILURE in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39ba98fbea7d430093fe87bab51d0386 : FAILURE in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/562115f11a1d4de3851d7afffbe13d9c : FAILURE in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5cb64e587e407aa8f1e662a859926b : FAILURE in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e43530c0d7441638cfb5d5b86c60b3c : SUCCESS in 3m 55s\n- cross-nova-tox-py38 https://zuul.opendev.org/t/openstack/build/ef2f00d2d3eb4734b2da04217c83acd6 : FAILURE in 7m 26s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/a6d6e3c20d394d068eff2e71540b318d : FAILURE in 7m 57s\n- cross-neutron-tox-py38 https://zuul.opendev.org/t/openstack/build/2ba26dbb56114f92a3627fc6a3feebed : FAILURE in 6m 22s","accounts_in_message":[],"_revision_number":3},{"id":"6876ee29c33e49aaa6f909e71e20424c26c59244","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-25 17:59:46.000000000","message":"Patch Set 3: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6513f4effc23357801dcfa53fde5a3567ff868fe","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-27 13:31:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1886fa0371d9db0a446cdcf5ecf35a0c7aa6afd4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-07-18 20:00:34.000000000","message":"Abandoned\n\ndefault values are proposed to change in new change and after that we do not need to suppress the warnings so this change is not needed - https://review.opendev.org/c/openstack/oslo.policy/+/924283","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b264413de89c628256ec59b2b6f6c04a5f576286","revisions":{"b1b12da6ad3cf469ee1bfb488a2e9e5364cf7d2e":{"kind":"REWORK","_number":1,"created":"2021-07-05 19:16:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/39/799539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/39/799539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/1"}}},"commit":{"parents":[{"commit":"0bc8a2e70faaf8600a44c6e7e366d2296adc128b","subject":"Merge \"Replace getargspec with getfullargspec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/0bc8a2e70faaf8600a44c6e7e366d2296adc128b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-05 18:01:42.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-05 19:15:52.000000000","tz":-300},"subject":"Make policy rule default change warning configurable","message":"Make policy rule default change warning configurable\n\nAll the services policies are moving to new secure RBAC\n(scope and new defaults), every policy rule is changing\ntheir default value.\nExample: https://review.opendev.org/c/openstack/nova/+/717175/11/nova/policies/server_groups.py#27\n\noslo policy log the warnings if the default value of policy rule\n(if not overridden) is changed, so there are warnings for\nevery policy rule on every API request, everytime policy\nis initialized which end up a lot of warnings (thousands)\nin log.\n\nCurrently, warnings can be disabled via a hardcoded\nflag \u0027suppress_default_change_warnings\u0027 especially for\nunit/functional test run.\n\nBut there is no way for the operator to disable/enable these\nwarnings.\n\nThis commit adds a new config option, \u0027warn_default_change\u0027\nto control these warnings. NOTE this config only controls the\ndefault change warning, not if policy name is changed.\n\nNOTE: default value of this config option is False, which\nmeans do not log warning by default. If the operator wants,\nthey can enable it. This is what many projects are doing\ncurrently (disabled warnings via \u0027suppress_default_change_warnings\u0027)\n\n- https://codesearch.opendev.org/?q\u003dsuppress_default_change_warnings\u0026i\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n\nChange-Id: I9f35db891d9a0812a4aa0b538505f6ef65ecd0d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b1b12da6ad3cf469ee1bfb488a2e9e5364cf7d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b1b12da6ad3cf469ee1bfb488a2e9e5364cf7d2e"}]},"branch":"refs/heads/master"},"150fe210d8fc44177abfcaf3496fe52d3ee9a6b2":{"kind":"REWORK","_number":2,"created":"2021-07-12 17:42:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/39/799539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/39/799539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/2"}}},"commit":{"parents":[{"commit":"0bc8a2e70faaf8600a44c6e7e366d2296adc128b","subject":"Merge \"Replace getargspec with getfullargspec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/0bc8a2e70faaf8600a44c6e7e366d2296adc128b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-05 18:01:42.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-07-12 17:42:18.000000000","tz":0},"subject":"Make policy rule default change warning configurable","message":"Make policy rule default change warning configurable\n\nAll the services policies are moving to new secure RBAC\n(scope and new defaults), every policy rule is changing\ntheir default value.\nExample: https://review.opendev.org/c/openstack/nova/+/717175/11/nova/policies/server_groups.py#27\n\noslo policy log the warnings if the default value of policy rule\n(if not overridden) is changed, so there are warnings for\nevery policy rule on every API request, everytime policy\nis initialized which end up a lot of warnings (thousands)\nin log.\n\nCurrently, warnings can be disabled via a hardcoded flag\n\u0027suppress_default_change_warnings\u0027 especially for unit/functional\ntest run. And there is no way for the operator to disable/enable\nthese warnings.\n\nThis commit adds a new config option, \u0027warn_default_change\u0027\nto control these warnings. NOTE this config only controls the\ndefault change warning, not if policy name is changed.\n\nNOTE: During the migration to new secure RBAC, default value of this\nconfig option is False, which means do not log warning by default.\nIf the operator wants, they can enable it. This is what many projects are doing\ncurrently (disabled warnings via \u0027suppress_default_change_warnings\u0027) so\nnot changning what operator gets currently.\n\n- https://codesearch.opendev.org/?q\u003dsuppress_default_change_warnings\u0026i\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n\nChange-Id: I9f35db891d9a0812a4aa0b538505f6ef65ecd0d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/150fe210d8fc44177abfcaf3496fe52d3ee9a6b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/150fe210d8fc44177abfcaf3496fe52d3ee9a6b2"}]},"branch":"refs/heads/master"},"b264413de89c628256ec59b2b6f6c04a5f576286":{"kind":"REWORK","_number":3,"created":"2022-04-25 13:57:40.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/39/799539/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/39/799539/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/39/799539/3"}}},"commit":{"parents":[{"commit":"9bc1783400f9960b7132bf631c690a98fbfc8066","subject":"Don\u0027t raise InvalidScope exception when do_raise\u003dFalse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/9bc1783400f9960b7132bf631c690a98fbfc8066"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-05 18:01:42.000000000","tz":-300},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2022-04-25 13:57:37.000000000","tz":120},"subject":"Make policy rule default change warning configurable","message":"Make policy rule default change warning configurable\n\nAll the services policies are moving to new secure RBAC\n(scope and new defaults), every policy rule is changing\ntheir default value.\nExample: https://review.opendev.org/c/openstack/nova/+/717175/11/nova/policies/server_groups.py#27\n\noslo policy log the warnings if the default value of policy rule\n(if not overridden) is changed, so there are warnings for\nevery policy rule on every API request, everytime policy\nis initialized which end up a lot of warnings (thousands)\nin log.\n\nCurrently, warnings can be disabled via a hardcoded flag\n\u0027suppress_default_change_warnings\u0027 especially for unit/functional\ntest run. And there is no way for the operator to disable/enable\nthese warnings.\n\nThis commit adds a new config option, \u0027warn_default_change\u0027\nto control these warnings. NOTE this config only controls the\ndefault change warning, not if policy name is changed.\n\nNOTE: During the migration to new secure RBAC, default value of this\nconfig option is False, which means do not log warning by default.\nIf the operator wants, they can enable it. This is what many projects are doing\ncurrently (disabled warnings via \u0027suppress_default_change_warnings\u0027) so\nnot changning what operator gets currently.\n\n- https://codesearch.opendev.org/?q\u003dsuppress_default_change_warnings\u0026i\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n\nChange-Id: I9f35db891d9a0812a4aa0b538505f6ef65ecd0d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b264413de89c628256ec59b2b6f6c04a5f576286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b264413de89c628256ec59b2b6f6c04a5f576286"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
