)]}'
{"id":"openstack%2Foslo.policy~801554","triplet_id":"openstack%2Foslo.policy~master~I73621ced00404d164fdb23f077ee36fbb6faf717","project":"openstack/oslo.policy","branch":"master","topic":"typo","hashtags":[],"change_id":"I73621ced00404d164fdb23f077ee36fbb6faf717","subject":"Fix a typo in the document","status":"MERGED","created":"2021-07-21 02:30:01.000000000","updated":"2021-07-22 09:35:39.000000000","submitted":"2021-07-22 09:34:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801554-typo","meta_rev_id":"b07193cb1720d15eab34c4e63551afeba047b723","_number":801554,"virtual_id_number":801554,"owner":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 09:34:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-07-22 09:02:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-07-22 09:02:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-21 02:59:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 09:02:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"8beefdbac1a9c2aeaf168d673a7373218d9e443a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2021-07-21 02:30:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5576728a3821feaa37966bd41a6b4e32f68ff505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 02:59:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5636f3ba0bfc41f782633adfc4508482 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc841b892954363931620dee279fcca : SUCCESS in 3m 50s\n- cross-nova-tox-py38 https://zuul.opendev.org/t/openstack/build/6f7b217bf6c2421ea931c1f9b3a0547f : SUCCESS in 12m 16s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/4f90d0b4d5f54b4dbb0a2348f6198e1e : SUCCESS in 19m 31s\n- cross-neutron-tox-py38 https://zuul.opendev.org/t/openstack/build/64ad6e00ec934e47bb07052b6e71b81e : SUCCESS in 24m 11s","accounts_in_message":[],"_revision_number":1},{"id":"44be50067a2b8185569d198cc4f9fddf2b170321","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-22 09:02:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial fix. LGTM","accounts_in_message":[],"_revision_number":1},{"id":"3534f8240c2f28ed5305ae4540f53bfe4679b514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 09:03:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1376cee51ed1b7a75a29361fcd033709cd2a2327","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 09:34:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01fe3222a0404fd6ace28e83bbeabfbb : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb10b5e9884148f3b5053faa3572d101 : SUCCESS in 4m 48s\n- cross-nova-tox-py38 https://zuul.opendev.org/t/openstack/build/08cdeda890944e479032dcfced74d445 : SUCCESS in 13m 47s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/b871c14fc32c40438403beb993fbd200 : SUCCESS in 19m 41s\n- cross-neutron-tox-py38 https://zuul.opendev.org/t/openstack/build/6e48925382f4491baade354760675b69 : SUCCESS in 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e2a99f7582c5ae34b379cf48df0f270d7f07900b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 09:34:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b07193cb1720d15eab34c4e63551afeba047b723","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 09:35:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6df86a260a0481ca505389ef89a478b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7fd9f4fcd43fb78534921530d981634ec516344","revisions":{"c7fd9f4fcd43fb78534921530d981634ec516344":{"kind":"REWORK","_number":1,"created":"2021-07-21 02:30:01.000000000","uploader":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"ref":"refs/changes/54/801554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/54/801554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/54/801554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/54/801554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/54/801554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/54/801554/1"}}},"commit":{"parents":[{"commit":"fa5f76e55910040f9132bd19940ceee1307ff303","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/fa5f76e55910040f9132bd19940ceee1307ff303"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2021-07-21 02:27:42.000000000","tz":480},"committer":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2021-07-21 02:29:57.000000000","tz":480},"subject":"Fix a typo in the document","message":"Fix a typo in the document\n\nThis patch changes \u0027oslopolicy-policy-generator\u0027 to\n\u0027oslopolicy-checker\u0027 in oslopolicy-checker.rst.\n\nChange-Id: I73621ced00404d164fdb23f077ee36fbb6faf717\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/c7fd9f4fcd43fb78534921530d981634ec516344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/c7fd9f4fcd43fb78534921530d981634ec516344"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
