)]}'
{"id":"openstack%2Foslo.policy~855754","triplet_id":"openstack%2Foslo.policy~master~Id1d4120218744371aed127559ac6d5abf0778648","project":"openstack/oslo.policy","branch":"master","topic":"release-health-check","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-07 09:27:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id1d4120218744371aed127559ac6d5abf0778648","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:37:09.000000000","updated":"2022-09-07 09:27:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bdf92cc25be11deea16f54a377bac12cea0200c2","_number":855754,"virtual_id_number":855754,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-03 22:11:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 22:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cba9151b9b104429455af2b307ce48172a9a9c84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:37:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be0b860a30efc1d9d633c783ddc9b32dfa175eeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 22:11:55.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48689b883596434b8feec28b04831577\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd7b5bea811d41b19365ba513a21e60d : SUCCESS in 1h 32m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6fdecca9b47490caaed9cbf9b7ac231 : FAILURE in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/841625c863ae410bbc736552428c6694 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f9444877a654776a0c62357a4a3dfb8 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/099b17f4e5e74aa19c99ec798dc42273 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ab55071c3f747f3ab87d68a9e0880b0 : SUCCESS in 4m 34s\n- cross-nova-tox-py38 https://zuul.opendev.org/t/openstack/build/460453147d254015a89db708153e4651 : SUCCESS in 13m 01s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/91bc29f4d1f54f2890a76e35bd55726b : SUCCESS in 19m 45s\n- cross-neutron-tox-py38 https://zuul.opendev.org/t/openstack/build/23ac5ca9b30847f9b4c015fc14ae73ea : SUCCESS in 23m 38s","accounts_in_message":[],"_revision_number":1},{"id":"bdf92cc25be11deea16f54a377bac12cea0200c2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:27:30.000000000","message":"Abandoned\n\nGate check passed (pep8 job failure is related to the health check patch), no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05fcd6a42c1bb3141eb0dd8518550c73cf3d35b8","revisions":{"05fcd6a42c1bb3141eb0dd8518550c73cf3d35b8":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:37:09.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/54/855754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/54/855754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/54/855754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/54/855754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/54/855754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/54/855754/1"}}},"commit":{"parents":[{"commit":"5bd767be790f21d6ec0af0589f50137b207d94e8","subject":"Fix generation of sample policy files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/5bd767be790f21d6ec0af0589f50137b207d94e8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:36:55.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:37:07.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: Id1d4120218744371aed127559ac6d5abf0778648\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/05fcd6a42c1bb3141eb0dd8518550c73cf3d35b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/05fcd6a42c1bb3141eb0dd8518550c73cf3d35b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
