)]}'
{"id":"openstack%2Foslo.policy~921913","triplet_id":"openstack%2Foslo.policy~master~I0babc058c2e17354b55a98a873fe95da5da4d62d","project":"openstack/oslo.policy","branch":"master","topic":"fix-http-check-no-rule-name","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2026-04-21 09:28:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0babc058c2e17354b55a98a873fe95da5da4d62d","subject":"Fix rule check missing rule name","status":"ABANDONED","created":"2024-06-13 06:59:17.000000000","updated":"2026-04-21 09:28:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e3b64f108412ee2dc2a29e05bc602b6b780df389","_number":921913,"virtual_id_number":921913,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-13 16:22:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-13 09:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a140b6a97c5dc2ba13c297d5747e962d80fad06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-06-13 06:59:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4191ef5a5415cf8fef048849058441f709b2dbda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 09:00:11.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ad0386246314a5bb7dd0e53a9af8524\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e75005b57d84c26a65c86223284c635 : SUCCESS in 1h 55m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d30bbdfb7ba4475ae40295584cfdb9f : FAILURE in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fec004e9c214de4900d89c64d15f30d : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/346ee6bdb37947d0a5e5146260eed2c5 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/619a304597e644aeaac88bbde780c5c2 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ec0b4118824a3fa0d38c325ec20fe4 : SUCCESS in 3m 11s\n- cross-nova-tox-py310 https://zuul.opendev.org/t/openstack/build/2144d00db896446ca15a9f2ea6140260 : SUCCESS in 14m 05s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/62aa2571db8f4300867a6ab8047c034c : SUCCESS in 27m 18s\n- cross-neutron-tox-py310 https://zuul.opendev.org/t/openstack/build/906b79c878e243e88f40ef01e42d9159 : SUCCESS in 31m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d192813e44da614418384b860ca6318c9a4dd670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-06-13 14:28:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7499d02cbce13033bdc16ae24d6c6b707929be66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 16:22:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/859e93157cd248d69f413d24914b3d58\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47f6763abf4c4ae0836a7c8f090eef05 : SUCCESS in 1h 48m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f90822666a74958aad6a0af6e0daea7 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6b89cc0872442239243a5fc6fd0993a : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e854da9af8f431ba3706358eadf5563 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/989cd63a2d15490aafa597979339f9a5 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d1ae6ed64a43ec9907bf99ce98d51d : SUCCESS in 3m 04s\n- cross-nova-tox-py310 https://zuul.opendev.org/t/openstack/build/74ea41131d5c4c068d27b1d3c07d6494 : SUCCESS in 12m 07s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/31ec1b1260ca43128aa831fefe0810d2 : SUCCESS in 25m 17s\n- cross-neutron-tox-py310 https://zuul.opendev.org/t/openstack/build/805a234eff7843f48d94af7f104bbdac : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e3b64f108412ee2dc2a29e05bc602b6b780df389","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2026-04-21 09:28:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b9d4597e51f87daae1d52f57aa5a579399571e36","revisions":{"b857b7444689a8d3c4227fc0cef5acbc16c5124e":{"kind":"REWORK","_number":1,"created":"2024-06-13 06:59:17.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/13/921913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/13/921913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/921913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/921913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/921913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/13/921913/1"}}},"commit":{"parents":[{"commit":"c7083ec2c591e0f921d3532135e235695df0778f","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/c7083ec2c591e0f921d3532135e235695df0778f"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-06-13 06:55:24.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-06-13 06:59:14.000000000","tz":480},"subject":"Fix rule check missing rule name","message":"Fix rule check missing rule name\n\nWhen running rule check, rule name was missing for checks like\nHTTPCheck.\n\nThis make it hard for remote http policy check server to optimize as no\nrule information is provided.\n\nThis patch make sure we give correct rule name for RuleCheck.\n\nChange-Id: I0babc058c2e17354b55a98a873fe95da5da4d62d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b857b7444689a8d3c4227fc0cef5acbc16c5124e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b857b7444689a8d3c4227fc0cef5acbc16c5124e"}]},"branch":"refs/heads/master"},"b9d4597e51f87daae1d52f57aa5a579399571e36":{"kind":"REWORK","_number":2,"created":"2024-06-13 14:28:31.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/13/921913/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/13/921913/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/921913/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/921913/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/13/921913/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/13/921913/2"}}},"commit":{"parents":[{"commit":"c7083ec2c591e0f921d3532135e235695df0778f","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/c7083ec2c591e0f921d3532135e235695df0778f"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-06-13 06:55:24.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-06-13 14:28:19.000000000","tz":480},"subject":"Fix rule check missing rule name","message":"Fix rule check missing rule name\n\nWhen running rule check, rule name was missing for checks like\nHTTPCheck.\n\nThis make it hard for remote http policy check server to optimize as no\nrule information is provided.\n\nThis patch make sure we give correct rule name for RuleCheck.\n\nChange-Id: I0babc058c2e17354b55a98a873fe95da5da4d62d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b9d4597e51f87daae1d52f57aa5a579399571e36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/b9d4597e51f87daae1d52f57aa5a579399571e36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
