)]}'
{"id":"openstack%2Foslo.policy~994271","triplet_id":"openstack%2Foslo.policy~master~Id8678542743b743e08a5578739468ae7215ed887","project":"openstack/oslo.policy","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-22 22:26:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id8678542743b743e08a5578739468ae7215ed887","subject":"Do not install code to build release notes","status":"MERGED","created":"2026-06-22 12:27:04.000000000","updated":"2026-06-22 22:27:00.000000000","submitted":"2026-06-22 22:26:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994271","meta_rev_id":"80bf185fa2d1a627e38fa3a32bea4a64333f58df","_number":994271,"virtual_id_number":994271,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-22 22:26:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-22 19:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-22 19:26:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-22 14:24:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-22 19:26:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"229de1c7ab3afdb6b1331962bf5ce33d2df2f110","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-22 12:27:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c76250f5b671105122c4bf39e19590569189c2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:24:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f212e7a7d7fc4f18ad93fcbdddca010b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c113e598c441441e8e1a5b9610fd5484 : SUCCESS in 1h 51m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f56da7be778648d4bd4dd2c3f307a5d0 : FAILURE in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5713518e86e4310b09f995eb200bcad : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cc995ef7d5f4a59ba44ac1ff2f839ce : SUCCESS in 2m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8850c537a9e4e9f92cee0cd7966c5e0 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d30a04fc4c04ba3bd81afdd47297bc9 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aee733b1f1954ae6b1cebf5b969b5db7 : SUCCESS in 3m 11s\n- cross-nova-tox-py313 https://zuul.opendev.org/t/openstack/build/0f1c9f0816794e02b587aedf849dfa31 : SUCCESS in 13m 54s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/00848d1c558d41f787b2b951111ba49d : SUCCESS in 28m 59s\n- cross-neutron-tox-py313 https://zuul.opendev.org/t/openstack/build/5e36b63d270e41c6b206908a4ab04e3f : SUCCESS in 19m 03s\n\nWarning:\n  Comments left for invalid file oslo_policy/generator.py\n  Comments left for invalid file oslo_policy/policy.py","accounts_in_message":[],"_revision_number":1},{"id":"41d985239fcd2306c063bf0b82102cee48aefb3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-22 19:26:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"bac12cfd3095d53908dc5727850aae83498840f4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-22 19:26:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3ac125a7a313768353e704b169fcc62db12fbb20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 21:00:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad7ff70a95d40f1a9f0aed8e680d705\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f42a654f81bd4761a94e6308bfb20f73 : SUCCESS in 1h 14m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e51810a39e14712b17904306fe82c49 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d179c635e655476f936538c49a8517c0 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/519bb5b104364c3cabde86611fbdc66f : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8183a70126694a5999eedf5136344b84 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e351ae8805c4b7e8cb59843b7f8c248 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54c540a604624ef6b06c1cf2624029a6 : SUCCESS in 3m 00s\n- cross-nova-tox-py313 https://zuul.opendev.org/t/openstack/build/60a91c376fd446de862b00ea539bc4aa : SUCCESS in 10m 31s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/2b538e2a6cab4c39a571c5d2a93f8346 : SUCCESS in 24m 17s\n- cross-neutron-tox-py313 https://zuul.opendev.org/t/openstack/build/6acb6ba7f60b402c90982d9e35bb682a : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":2},{"id":"9c081fa55bb032558b35478cb124d8e33af3f329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 21:00:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14ea02039aa7a01f5f1549766520e7b4907e2a30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 22:26:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1146686f2eb94142a50dffdfe2850744\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ed1e97c39a74d4f9d9ab30ff22424b1 : SUCCESS in 1h 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d0f243bc96d4ce4b3e6f1310495de1b : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32f4d67645d349c7ade067263cc2a93f : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ab2db8e8af649deafe347ad3a28eeaf : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12b4ca49bf9b4766997b1e4015fc891b : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c842abac0ef74d34b1dde52da912141c : SUCCESS in 2m 59s\n- cross-nova-tox-py313 https://zuul.opendev.org/t/openstack/build/8444f960b0b840adbc60c724d20f0a0e : SUCCESS in 14m 21s\n- cross-nova-tox-functional https://zuul.opendev.org/t/openstack/build/b73ca6e6c4ff4039b7bbf50792130b04 : SUCCESS in 27m 17s\n- cross-neutron-tox-py313 https://zuul.opendev.org/t/openstack/build/fb03cd50793243d3b5859371aa77ddfe : SUCCESS in 22m 45s","accounts_in_message":[],"_revision_number":2},{"id":"9788dd2ffe90265f4786a2452c1d0c228ee4d9fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 22:26:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"80bf185fa2d1a627e38fa3a32bea4a64333f58df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 22:27:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b46e4e1db7b149cab6b02168d9d007a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ce871c3d8841a4b31201cf2e475813 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0cd3666b7fe46a393a965bb0adf61a8 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e81c02f4ca70209aed0a3cf4457dde603216fa9","revisions":{"1422f22c0f9a1e87ca6389c3625525f8942603d0":{"kind":"REWORK","_number":1,"created":"2026-06-22 12:27:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/994271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/71/994271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/71/994271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/71/994271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/71/994271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/71/994271/1"}}},"commit":{"parents":[{"commit":"fd73c1a90cfd9291d7a3ccfea50920eae3d93302","subject":"Deprecate version module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/fd73c1a90cfd9291d7a3ccfea50920eae3d93302"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 12:26:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 12:26:53.000000000","tz":540},"subject":"Do not install code to build release notes","message":"Do not install code to build release notes\n\n... because we require only doc requirements.\n\nAlso remove explicit usage of requirements.txt because it is now\nautomatically detected and installed.\n\nChange-Id: Id8678542743b743e08a5578739468ae7215ed887\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/1422f22c0f9a1e87ca6389c3625525f8942603d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/1422f22c0f9a1e87ca6389c3625525f8942603d0"}]},"branch":"refs/heads/master"},"1e81c02f4ca70209aed0a3cf4457dde603216fa9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-22 19:26:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/994271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.policy","ref":"refs/changes/71/994271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/71/994271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/71/994271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.policy refs/changes/71/994271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.policy refs/changes/71/994271/2"}}},"commit":{"parents":[{"commit":"283fc59476334eb9cd3ba455a438629c579700fe","subject":"typing: Fix compatibility with typed oslo.config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/283fc59476334eb9cd3ba455a438629c579700fe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 12:26:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 19:26:43.000000000","tz":0},"subject":"Do not install code to build release notes","message":"Do not install code to build release notes\n\n... because we require only doc requirements.\n\nAlso remove explicit usage of requirements.txt because it is now\nautomatically detected and installed.\n\nChange-Id: Id8678542743b743e08a5578739468ae7215ed887\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/1e81c02f4ca70209aed0a3cf4457dde603216fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.policy/commit/1e81c02f4ca70209aed0a3cf4457dde603216fa9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"fd73c1a90cfd9291d7a3ccfea50920eae3d93302","ours":"1422f22c0f9a1e87ca6389c3625525f8942603d0","theirs":"283fc59476334eb9cd3ba455a438629c579700fe","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
