)]}'
{"id":"openstack%2Foslo.privsep~706921","triplet_id":"openstack%2Foslo.privsep~master~Ice4e7ef5b05c750e24a10ee1a133d45385341c54","project":"openstack/oslo.privsep","branch":"master","topic":"remove-outdated-headers","hashtags":[],"change_id":"Ice4e7ef5b05c750e24a10ee1a133d45385341c54","subject":"remove outdated header","status":"MERGED","created":"2020-02-10 17:33:40.000000000","updated":"2020-02-13 12:12:08.000000000","submitted":"2020-02-13 12:06:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706921-1581595614239-e503d632","meta_rev_id":"5b66ceb12c98b49c532dbae2dab0a005791e131a","_number":706921,"virtual_id_number":706921,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-02-12 10:16:58.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 12:06:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-13 10:27:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-13 10:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 14:39:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-13 10:27:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-13 12:06:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a56788261cba6c03f52828941183483c809ff7a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-10 17:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f687fb4d90f765d7289c7c75d7e22ced24468c1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 19:14:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b5b45502d26c49f2acf60f49693344d3 : SUCCESS in 5m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eea4e0668fb24664bcf22b06e9078162 : SUCCESS in 1h 32m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41e04a2563974109aad8b6bdbbac673b : SUCCESS in 1h 29m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7fd6c98f814404bac084689cb8860b3 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22efd8d121c442f5832fd8c8e8776c66 : SUCCESS in 7m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0aa84c44d8e456b82e7420438432ce0 : FAILURE in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/062607b08c224b299a48b77ace799837 : SUCCESS in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8471b71fc616474cb02318c8098cb3ec : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb7097863f964270a6022903fad25b07 : FAILURE in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89bd6b8eec684e8bb24842b875de1936 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d13a00c758e02309c6bc13e03785c63f1f56331a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-11 14:39:35.000000000","message":"Patch Set 1:\n\nNeeds https://review.opendev.org/#/c/705715/","accounts_in_message":[],"_revision_number":1},{"id":"edf6299c3f8c6576a1ea5646f726164362295708","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-11 17:52:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"864254decdd23d1fe983fda847b744150a269902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 18:40:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bc0fccdefafc4030ab80ed81246a0e00 : RETRY_LIMIT in 2m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d577937ba2384e24b06d6264b8224e73 : RETRY_LIMIT in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03037967d2f54d4886b8f82c9d58ce93 : RETRY_LIMIT in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88043b15b6c84aa485354c15cb71915d : RETRY_LIMIT in 2m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3e3088bc88d4157a2c5c6a4ae2bfc1d : RETRY_LIMIT in 2m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3154b9856ae54a07af7a4463cf201627 : RETRY_LIMIT in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8767bb216c154f04addc0554bfdd5acb : RETRY_LIMIT in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3623e51347da4a8f86412598efd4ce6b : RETRY_LIMIT in 2m 33s","accounts_in_message":[],"_revision_number":2},{"id":"d392e0c1ae39b45d25eb2660e52a992b3b2082e4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-12 10:16:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ae05ef2347b38f687cc522e765d849b5bea3cc35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:07:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0d89cd96a84f4e5ba0bcac36f6e759c0 : SUCCESS in 3m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cc19a5aa3014a8daf088d88ab181102 : SUCCESS in 1h 47m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4441db255e04410a48b19f21ae0e20a : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/447509fc76ea4e9e84d7715fc80d79ed : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ec1dec3bca344349cf81b2ea59df1e0 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10c64bfbc52c4fdb906939a456b9f082 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a385933f0e4c40efb5b06bed6c46e6be : FAILURE in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b18d4313c84423829c7f1c5ae6fa07 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a0a48335d091d659a0616210be51362aa7adcec3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-13 10:27:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6f87d2a3e6717ea4243eceae9db9bf331f361cd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:37:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f97f37a0bf7620d93935bf5529fd40ad9eb70f32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:06:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"53535c8175e1208cd6a806108e07dd4c4a91329c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:06:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d60bcf76b65b4b57b3414400d0d65ba3 : SUCCESS in 4m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce7b828403bc4f7c908ec340bb2ab7ec : SUCCESS in 1h 22m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0ec37d50c92454e98519d54ebbe0dff : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a59918a8b27a411998eb11f95ab6cfb6 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98cbf292c2ca496ab5b23cc2682043a1 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65cdf3027b7f441592d99557b3a54089 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56f64f936e949359961cfd94f0fd182 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1ae5053a7b76ccc519d9f19c759be5e27c13473c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:12:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d84b40a00d4414b4feb37b82afc3c9 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"00d599017844dc0580c8fbdebdb2b272628291b6","revisions":{"0874333831c52227a80293d8a9960d2aef0579f3":{"kind":"REWORK","_number":1,"created":"2020-02-10 17:33:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/21/706921/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/21/706921/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/21/706921/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/21/706921/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/21/706921/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/21/706921/1"}}},"commit":{"parents":[{"commit":"a3037870e5dddd3917ee51eae98908e28adf3a94","subject":"Merge \"tox: Trivial cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/a3037870e5dddd3917ee51eae98908e28adf3a94"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-10 17:33:35.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-10 17:33:35.000000000","tz":60},"subject":"remove outdated header","message":"remove outdated header\n\nChange-Id: Ice4e7ef5b05c750e24a10ee1a133d45385341c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/0874333831c52227a80293d8a9960d2aef0579f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/0874333831c52227a80293d8a9960d2aef0579f3"}]},"branch":"refs/heads/master"},"00d599017844dc0580c8fbdebdb2b272628291b6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-11 17:52:11.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/21/706921/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/21/706921/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/21/706921/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/21/706921/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/21/706921/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/21/706921/2"}}},"commit":{"parents":[{"commit":"c4d72cfcdf57738644318bc792152428fd4c43b4","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c4d72cfcdf57738644318bc792152428fd4c43b4"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-10 17:33:35.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-11 17:51:53.000000000","tz":60},"subject":"remove outdated header","message":"remove outdated header\n\nChange-Id: Ice4e7ef5b05c750e24a10ee1a133d45385341c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/00d599017844dc0580c8fbdebdb2b272628291b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/00d599017844dc0580c8fbdebdb2b272628291b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
