)]}'
{"id":"openstack%2Foslo.privsep~725322","triplet_id":"openstack%2Foslo.privsep~master~I4f1501b2a7dc40fc8425b6d6443b429b064a0b4b","project":"openstack/oslo.privsep","branch":"master","hashtags":[],"change_id":"I4f1501b2a7dc40fc8425b6d6443b429b064a0b4b","subject":"Align contributing doc with oslo\u0027s policy","status":"MERGED","created":"2020-05-04 16:27:56.000000000","updated":"2020-06-30 18:04:32.000000000","submitted":"2020-06-30 18:00:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725322-1593540035914-2d1bbc0e","meta_rev_id":"2d72427c6ca1e8dd7355266699a95a7a31266c69","_number":725322,"virtual_id_number":725322,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 18:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-06-30 15:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-06-30 15:54:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 15:54:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-06-30 18:00:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c3baae1569a21ba19ca910d60be593531c051a9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-04 16:27:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d31b10a0ff14073d721fc526e2dc137bbe7d01e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:34:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77819c08b750427ebb2378c54e5aa955 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29f2688d7c9745a887f30e4fa78bd9a7 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3d48dd816a7cf07205ea48abc927e18710f0d1b2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-30 15:54:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial doc change. Fast approving","accounts_in_message":[],"_revision_number":1},{"id":"2d46ed672fccf7b270ac797606e2099ab81ac808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 15:54:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33dce33ca79db1e5c874610deed1b27f92fd1ff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 16:09:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b8f89d232384494bf17888ddecd9718 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939abc61cbce4b889066f3dbd111142a : POST_FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"91c6e67d53c400d6f917d717e89dc23ce36cc892","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-30 17:39:08.000000000","message":"Patch Set 1:\n\nRECHECK","accounts_in_message":[],"_revision_number":1},{"id":"5fd09f25b9528852f7ff0e2dba755f29413e957f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 17:49:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a2222f71d7d4cc2a1fc00ce953df916 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e614fda18854e0fb9724cf2e238355c : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"47068e86a780969d426b552ddd1f4180c70b0e35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 17:49:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad5dacfaf1adc448f3136e47a1e35908a6e138cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 18:00:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2faa78861c62c734ab19e9098c5c7b223547c701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 18:00:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10a547a3517744d483d254905e299f3b : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c454483801345bca4e9569c7d713eb5 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2d72427c6ca1e8dd7355266699a95a7a31266c69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 18:04:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c487970b6c74f229b793dbbc67f1683 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c38835499cb2172da87b1ca749f38a6dfdf7b922","revisions":{"c38835499cb2172da87b1ca749f38a6dfdf7b922":{"kind":"REWORK","_number":1,"created":"2020-05-04 16:27:56.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/22/725322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/22/725322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/22/725322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/22/725322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/22/725322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/22/725322/1"}}},"commit":{"parents":[{"commit":"5f04cd8dd67ddc028113d7625d9d029891c95f82","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/5f04cd8dd67ddc028113d7625d9d029891c95f82"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-04 16:27:35.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-04 16:27:35.000000000","tz":120},"subject":"Align contributing doc with oslo\u0027s policy","message":"Align contributing doc with oslo\u0027s policy\n\nRelated to:\n- https://review.opendev.org/#/c/723044/\n- https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nChange-Id: I4f1501b2a7dc40fc8425b6d6443b429b064a0b4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c38835499cb2172da87b1ca749f38a6dfdf7b922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c38835499cb2172da87b1ca749f38a6dfdf7b922"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
