)]}'
{"id":"openstack%2Foslo.privsep~728513","triplet_id":"openstack%2Foslo.privsep~master~Idb12608777b8f81747a3d93a06f332057a4871bf","project":"openstack/oslo.privsep","branch":"master","hashtags":[],"change_id":"Idb12608777b8f81747a3d93a06f332057a4871bf","subject":"Remove the unused coding style modules","status":"MERGED","created":"2020-05-15 17:10:37.000000000","updated":"2020-05-26 13:49:50.000000000","submitted":"2020-05-26 13:47:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728513-1590500852858-3b9071a1","meta_rev_id":"256f01df4df1b795a99cbcd9cfc6461dc247e9f4","_number":728513,"virtual_id_number":728513,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-05-17 02:58:14.000000000","_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 13:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-26 12:14:46.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2020-05-25 14:41:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-26 12:14:46.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 14:41:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-26 12:14:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-26 13:47:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46054c6b51e91e45be7d21a20f52966a433db46e","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-15 17:10:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5aebd41afff08d550658080c73ea4b1eb03207b","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-15 17:19:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"64f7620916afc83ce4dbe91959f4a0fb3c092493","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-17 02:58:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"623a1c98f12602400c67382efca778482637e5e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 04:19:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/54a8532e425b4c73ae66e48a80850698 : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73885c263503457dba817b1a94ee34b3 : SUCCESS in 1h 19m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f3f1b9143a14813aaec58cb053991dd : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7f14975ebe4bec826553a3c68dc839 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/120d7940f6bb46779c6091b0fe43e5d1 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47c4a2103b934b2da116fb077336bc9b : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249ddc4c8e8d439caf407844515e2dee : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d36ac7ae22d0e59d982f7e58c6a562d1a468d7fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-25 14:41:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"da308cbcbf460210d82c4565b97a3d61794352da","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-26 12:14:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c03d5e68b9fa94d63cda0c2e635a3d3d5479ba27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 12:14:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c2f4ad745e266da211502a3672cdc1dc088071b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 13:47:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/139b7c186992488288a577d3268d4e12 : SUCCESS in 5m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ab7f20abda047a49cc61fc82b63bbe2 : SUCCESS in 1h 26m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45e5c04734fa4ba9b233c181554454db : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29fcef2d81d246b492d502143a87ee29 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b7ba3736a004bc180b24b81314a978d : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/452fd55c4ab24161b509720a8a903572 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac2a4dbd962412ba779050b42ef6dc8 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":2},{"id":"504b993eb2150ad442e6522c6a03e74dd41bd13d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 13:47:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"256f01df4df1b795a99cbcd9cfc6461dc247e9f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 13:49:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f2235feb7b54df38c12efafa13d91f7 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"95d8f6d60a9baca0cbb8d07728a709ad49c6f2fa","revisions":{"c7b525103bd1554b8240d900e02057e8511d203c":{"kind":"REWORK","_number":1,"created":"2020-05-15 17:10:37.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/13/728513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/13/728513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/728513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/728513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/728513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/13/728513/1"}}},"commit":{"parents":[{"commit":"5f04cd8dd67ddc028113d7625d9d029891c95f82","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/5f04cd8dd67ddc028113d7625d9d029891c95f82"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 17:10:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 17:10:09.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nChange-Id: Idb12608777b8f81747a3d93a06f332057a4871bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c7b525103bd1554b8240d900e02057e8511d203c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c7b525103bd1554b8240d900e02057e8511d203c"}]},"branch":"refs/heads/master"},"95d8f6d60a9baca0cbb8d07728a709ad49c6f2fa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-15 17:19:28.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/13/728513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/13/728513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/728513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/728513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/728513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/13/728513/2"}}},"commit":{"parents":[{"commit":"5f04cd8dd67ddc028113d7625d9d029891c95f82","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/5f04cd8dd67ddc028113d7625d9d029891c95f82"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 17:10:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 17:19:25.000000000","tz":0},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\nmore info: https://github.com/openstack/requirements/blob/master/blacklist.txt\n\nChange-Id: Idb12608777b8f81747a3d93a06f332057a4871bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/95d8f6d60a9baca0cbb8d07728a709ad49c6f2fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/95d8f6d60a9baca0cbb8d07728a709ad49c6f2fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
