)]}'
{"id":"openstack%2Foslo.privsep~873513","triplet_id":"openstack%2Foslo.privsep~master~I5d66fccd9ffb07df0c2e4435ec3da767b3b61117","project":"openstack/oslo.privsep","branch":"master","topic":"fix/setgid","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-07-14 13:40:25.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-10 04:05:13.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-07-14 15:22:09.000000000","reason":"Change was submitted"},"35676":{"account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"last_update":"2023-07-14 15:22:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d66fccd9ffb07df0c2e4435ec3da767b3b61117","subject":"setgid should be called before setuid","status":"MERGED","created":"2023-02-13 09:52:50.000000000","updated":"2023-07-14 15:23:11.000000000","submitted":"2023-07-14 15:22:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873513-fix/setgid","meta_rev_id":"2263e2949e00bc2c301c378b74e71f9f3668839d","_number":873513,"virtual_id_number":873513,"owner":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-14 15:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},"all":[{"value":1,"date":"2023-03-03 08:47:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},{"value":2,"date":"2023-07-14 13:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-04-10 04:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},{"value":1,"date":"2023-07-14 13:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 10:38:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-02-13 11:35:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-02-13 13:06:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 12:09:42.000000000","updated_by":{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},"reviewer":{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},"state":"REVIEWER"},{"updated":"2023-03-03 05:18:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-04-10 04:05:13.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-04 12:10:32.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"a91b031396a991640ff2cad8ffb039cbb99d43a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-02-13 09:52:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f333ae968933b907435b46f0c16b1c9e87c48687","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-02-13 10:26:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"70a7804c90608e719ea73ec6556b4ea345db9ddb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 10:38:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"865da5db24c013f71e9157e2f45ece253ab98ceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-02-13 11:26:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3c24122cf5d20be6d42d4812e470dc26da8359ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-02-13 11:33:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a184e89721b25ae0f87ac280e99956309c745bb","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-02-13 11:34:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a533eee53ddad52771128d40075eef66ec9413c8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-13 11:35:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ca5cb0cd0a240692701b475687a5552f9dc2c1c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:06:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a89877aaeb4306bb13f1076c2660b1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d7731cb50a84368a516da5ecd1163bc : SUCCESS in 1h 29m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae9833dbede41ba921653795b6d4856 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5566bc2e9d0341e79a7c337ef5c88d45 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a91a3f0c9524f5b99b5edd5bdff5e25 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf71a06e64994ce1a6b4aed0bc62d159 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cac844eba91490c828bbffe5c6d1ee9 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":4},{"id":"7b7f0688a7d475fcd739b2646a8bf5c408caa684","author":{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},"date":"2023-02-15 12:09:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"86b120bdc50e6b4c8c1dd0aac612136b2cae947b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-03 05:18:45.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5ec03ed6dffcee1691ff745825fc9f6d53a356f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-03-03 07:49:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"325fdb6170bf6e462fd9bca20cc73afeaf6d01b0","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-03-03 08:28:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4e7af6b25bf88886b63dc9a268a1ef8fc6a5065e","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-03-03 08:46:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c424d795b92d448a229dca1f21e6e212e4516b14","author":{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"},"date":"2023-03-03 08:47:51.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2e1bb07abe08fccc447ea5a6b121016576051acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 09:35:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/091fd21848f74ad5b5e57a49c6038043\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afb7085b11e142c1b19ecb612ab7b2a7 : SUCCESS in 1h 29m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab68eec3680e41f996a2046913b1b5d3 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7946d6abf6aa43b49747e56d3a03d71d : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcdd82eead884e89b83964ec0b217242 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ead3c585af48cf892c7c445b307712 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3fa8e4f2d454a8bac1fd68f47189f60 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":5},{"id":"fb6534ac64ccd12a3b92769ed6858e8a12915eb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 04:05:13.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"91dd580ff37b1ca775ff19d1bf00b1c75200e670","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-14 13:40:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"dcf29995bc3f3d9f1657a26c1071c7d700cdea7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 13:40:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a7fe28017b226bd6f0688de33a01e22077c7a25a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 15:22:09.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faafd7e5aa734f0c95501eb730f1c8e5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fecffc137775487ca489ca4968a40aa8 : SUCCESS in 1h 35m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ebee5d820e4c00bdf4a6e4abf90076 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0169cb24f354e8885c31470011c8164 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ad0db1ee6764fca8aad2224f9b8059d : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cacc0ae1d1754f31bc4af0e1407b6c75 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef7b4f0047a4e519bca06982f81cffa : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fca813ef7c74383bdf8c1b380e14df1 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":5},{"id":"6b42501f54172f42da5e263c1c64aa3a1294110a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 15:22:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2263e2949e00bc2c301c378b74e71f9f3668839d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 15:23:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c3c5aee526d4a27b73396d9ac32b922\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95eb52f02b6e45d6874742aae2fc4795 : SUCCESS in 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75e0221497d940b89055a1e9fa86992e : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"33fec1971b1c79a5445bea3520bbc8576bffcda2","revisions":{"6dc5ed3626ad4247322e47c2278a0e9de1b2017b":{"kind":"REWORK","_number":1,"created":"2023-02-13 09:52:50.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/13/873513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/13/873513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/1"}}},"commit":{"parents":[{"commit":"c2b6df05e0c8bd2a34b2b7db1935ea605062e22c","subject":"Setup logging without fixing evenlet logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c2b6df05e0c8bd2a34b2b7db1935ea605062e22c"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-02-13 09:45:19.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-02-13 09:45:19.000000000","tz":60},"subject":"setgid should be called before setuid","message":"setgid should be called before setuid\n\nIf you setuid to a non-zero value first(meaning you\u0027re no longer root),\nthen call setgroups, the effective uid of the process\nis now no longer root, meaning that the internal setgid call fails\n\nThis also removes the duplicated if loop\n\nChange-Id: I5d66fccd9ffb07df0c2e4435ec3da767b3b61117\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/6dc5ed3626ad4247322e47c2278a0e9de1b2017b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/6dc5ed3626ad4247322e47c2278a0e9de1b2017b"}]},"branch":"refs/heads/master"},"476d358c7b3bd2e38f892ba0bc0d38db20d7f7ac":{"kind":"REWORK","_number":2,"created":"2023-02-13 10:26:11.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/13/873513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/13/873513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/2"}}},"commit":{"parents":[{"commit":"c2b6df05e0c8bd2a34b2b7db1935ea605062e22c","subject":"Setup logging without fixing evenlet logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c2b6df05e0c8bd2a34b2b7db1935ea605062e22c"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-02-13 09:45:19.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-02-13 10:25:47.000000000","tz":60},"subject":"setgid should be called before setuid","message":"setgid should be called before setuid\n\nIf you setuid to a non-zero value first(meaning you\u0027re no longer root),\nthen call setgroups, the effective uid of the process\nis now no longer root, meaning that the internal setgid call fails\n\nThis also removes the duplicated if loop\n\nChange-Id: I5d66fccd9ffb07df0c2e4435ec3da767b3b61117\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/476d358c7b3bd2e38f892ba0bc0d38db20d7f7ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/476d358c7b3bd2e38f892ba0bc0d38db20d7f7ac"}]},"branch":"refs/heads/master"},"3d233cba6c6a3bf80ae3b785de22cb9046d74021":{"kind":"REWORK","_number":3,"created":"2023-02-13 11:26:33.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/13/873513/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/13/873513/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/3"}}},"commit":{"parents":[{"commit":"c2b6df05e0c8bd2a34b2b7db1935ea605062e22c","subject":"Setup logging without fixing evenlet logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c2b6df05e0c8bd2a34b2b7db1935ea605062e22c"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-02-13 09:45:19.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-02-13 11:26:06.000000000","tz":60},"subject":"setgid should be called before setuid","message":"setgid should be called before setuid\n\nIf you setuid to a non-zero value first(meaning you\u0027re no longer root),\nthen call setgroups, the effective uid of the process\nis now no longer root, meaning that the internal setgid call fails\n\nThis also removes the duplicated if loop\n\nCloses-Bug: #1628360\nChange-Id: I5d66fccd9ffb07df0c2e4435ec3da767b3b61117\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/3d233cba6c6a3bf80ae3b785de22cb9046d74021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/3d233cba6c6a3bf80ae3b785de22cb9046d74021"}]},"branch":"refs/heads/master"},"61a3cc6945f0c3294c8e955736d31f362b34b4cb":{"kind":"REWORK","_number":4,"created":"2023-02-13 11:33:11.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/13/873513/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/13/873513/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/4"}}},"commit":{"parents":[{"commit":"c2b6df05e0c8bd2a34b2b7db1935ea605062e22c","subject":"Setup logging without fixing evenlet logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c2b6df05e0c8bd2a34b2b7db1935ea605062e22c"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-02-13 09:45:19.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-02-13 11:32:56.000000000","tz":60},"subject":"setgid should be called before setuid","message":"setgid should be called before setuid\n\nIf you setuid to a non-zero value first(meaning you\u0027re no longer root),\nthen call setgroups, the effective uid of the process\nis now no longer root, meaning that the internal setgid call fails\n\nThis also removes the duplicated if loop\n\nCloses-Bug: #1628360\nChange-Id: I5d66fccd9ffb07df0c2e4435ec3da767b3b61117\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/61a3cc6945f0c3294c8e955736d31f362b34b4cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/61a3cc6945f0c3294c8e955736d31f362b34b4cb"}]},"branch":"refs/heads/master"},"33fec1971b1c79a5445bea3520bbc8576bffcda2":{"kind":"REWORK","_number":5,"created":"2023-03-03 07:49:35.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/13/873513/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.privsep","ref":"refs/changes/13/873513/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.privsep refs/changes/13/873513/5"}}},"commit":{"parents":[{"commit":"c2b6df05e0c8bd2a34b2b7db1935ea605062e22c","subject":"Setup logging without fixing evenlet logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/c2b6df05e0c8bd2a34b2b7db1935ea605062e22c"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-02-13 09:45:19.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-03-03 07:49:26.000000000","tz":60},"subject":"setgid should be called before setuid","message":"setgid should be called before setuid\n\nIf you setuid to a non-zero value first(meaning you\u0027re no longer root),\nthen call setgroups, the effective uid of the process\nis now no longer root, meaning that the internal setgid call fails\n\nThis also removes the duplicated if loop\n\nCloses-Bug: #1628360\nChange-Id: I5d66fccd9ffb07df0c2e4435ec3da767b3b61117\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/33fec1971b1c79a5445bea3520bbc8576bffcda2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.privsep/commit/33fec1971b1c79a5445bea3520bbc8576bffcda2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35408,"name":"Luca Czesla","email":"luca.czesla@mail.schwarz","username":"czesla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
