)]}'
{"id":"openstack%2Foslo.reports~1002001","triplet_id":"openstack%2Foslo.reports~master~I0c6b5472e52c22211a533583bbf32fe22ab928f6","project":"openstack/oslo.reports","branch":"master","topic":"tox-use-extras","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-24 10:55:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c6b5472e52c22211a533583bbf32fe22ab928f6","subject":"Remove duplicated optional dependencies from test requirements","status":"MERGED","created":"2026-08-24 03:18:35.000000000","updated":"2026-08-24 10:56:39.000000000","submitted":"2026-08-24 10:55:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002001-tox-use-extras","meta_rev_id":"679629901fced85ff1509ffeba770797f3126d52","_number":1002001,"virtual_id_number":1002001,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 10:55:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-24 09:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-24 09:01:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-24 04:34:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-24 09:01:11.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"84bafd53d3c929641dbcc05610189ee9dd4fdfcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 03:18:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afbccb22bb140e1559f790d896886ef2cdf9452e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 04:34:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe3b666964c04c268f1611fe54c2d23c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d363e0c3c65b4daaaec422e040f1a21f : SUCCESS in 2m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dbad065045b4991a6e0f83263bd4c12 : SUCCESS in 1h 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae63e03f067a4a04bb6c9a3da3328af8 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3b7350b7cb44576a0eef54e4fda2dfb : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2c2c49f92e14f39a32c0b4bd5638ba4 : SUCCESS in 2m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f9f807ea837c4b3caea8420069777b02 : SUCCESS in 7m 14s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3c51ee6b5bfc4da190d1d001d8c36ea7 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9c0338dc7249a8af696205fc564988 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5044b1416bb6439088264fa8431e6097 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f0eeb91e7449d4204ed28def7c62057caf0f0adf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-24 09:01:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"730fc809536edd693997553763cd3e44f1107bf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 09:01:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9f25b3f6355e590fdc8c24d37d4d83b83efacbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 10:55:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e27eced3fa4d7ea1359ecd00fca102\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cd31f5866e804fe6a64f1cb01bf3185d : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fc8514a573f4998be748495763aa0c3 : SUCCESS in 1h 48m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd9ecaeb6ed49c494cffbd80a0172c0 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49668ff606424c4bb1fd9c131893c662 : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b384db00d3c4605bd347a537ae80e92 : SUCCESS in 2m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/10548d4aee6e47df9dc975a2cdfd3058 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28545f9ad5614aa987725247057b277a : SUCCESS in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d43179b9f3b84cd8b56e713530475e8d : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d7209f32b13ac192f686318a3ba485d16cb0c034","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 10:55:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"679629901fced85ff1509ffeba770797f3126d52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 10:56:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dedc72fcf06a452482854054e5e3dcfb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f120c79630a46ffb39a83e4136a38cd : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/470b5f17dcc143f597a673bde063101e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0db530cd07f15212439fcb48b1f19c0d54f800f3","revisions":{"0db530cd07f15212439fcb48b1f19c0d54f800f3":{"kind":"REWORK","_number":1,"created":"2026-08-24 03:18:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/1002001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.reports","ref":"refs/changes/01/1002001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/01/1002001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/01/1002001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/01/1002001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.reports refs/changes/01/1002001/1"}}},"commit":{"parents":[{"commit":"d7774edecce0fc751f56d9ef7f54cb550c9e8e86","subject":"zuul: Use openstack-python3-next-jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/d7774edecce0fc751f56d9ef7f54cb550c9e8e86"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 03:17:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 03:18:24.000000000","tz":540},"subject":"Remove duplicated optional dependencies from test requirements","message":"Remove duplicated optional dependencies from test requirements\n\n... and use extras to define these, instead.\n\nChange-Id: I0c6b5472e52c22211a533583bbf32fe22ab928f6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/0db530cd07f15212439fcb48b1f19c0d54f800f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/0db530cd07f15212439fcb48b1f19c0d54f800f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
