)]}'
{"id":"openstack%2Foslo.reports~1007387","triplet_id":"openstack%2Foslo.reports~master~I747b75e347d70378965ecbb2959f5545960e88a8","project":"openstack/oslo.reports","branch":"master","topic":"py315","hashtags":[],"change_id":"I747b75e347d70378965ecbb2959f5545960e88a8","subject":"Declare Python 3.15 support","status":"MERGED","created":"2026-09-25 04:45:55.000000000","updated":"2026-09-25 17:04:44.000000000","submitted":"2026-09-25 17:03:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007387-py315","meta_rev_id":"8c6c2cd7ea2c6089157f20232cfb6fa0f582a57d","_number":1007387,"virtual_id_number":1007387,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 17:03:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-25 15:19:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-25 15:19:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 06:23:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9322e03a68d4e7d1d72c8cfcb699eb6194f911fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 04:45:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b152c82565fd84c0020309ece209e73c36d4a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 06:23:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab11283246d40d6b7a8e62effc866b6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b3e651956ae1424180a5529de967f04b : SUCCESS in 4m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3068b390eb2a43bda20916602457e9a3 : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8d4e85cb50c471092bce2d30732662b : SUCCESS in 1h 31m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88045ca23b154c5a82b5853e123beac4 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebc19022220b4682999fc10b3d47e303 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6eadaf5d753b42e6b29b76323b65c07d : SUCCESS in 5m 53s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9bb9a1b2f549482dbe96920333f77fb4 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b455086d7e4590999f3b5206cb07e4 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4684211ccbe44272afe162ab272f2d18 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b165eaa51766649350e8358320a7a4a05950cadb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 15:19:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea684052c93caae62d5a9c45f0cea65e97e05b05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 15:20:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"817f47f1c1cc471f0ac1beb7cd727fbd94324338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 17:03:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da26341be2524e3a82fb5141f208313b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/22a560a91df44774b02179f8f7d5763d : SUCCESS in 3m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4109c419b88c4f8c802e40d5262a6aa9 : SUCCESS in 2m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb8ef40bee8d44d7be79438f32bb203d : SUCCESS in 1h 37m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a844906287ad414b98b05ff0d95624b9 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4cb9ca07cca24b51974b41da900b9b13 : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dadfa6096b0d45f6ac82b2eeb2220c80 : SUCCESS in 5m 35s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/71a1b29440b44bca9e047349ac4c98cf : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1afcdd854e5d486aad022cdf49ce175d : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b106e07fd7f4d9089d33985ae5f9de6 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4cde787b355f37542c1120d644b6fdb101ddf54f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 17:03:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8c6c2cd7ea2c6089157f20232cfb6fa0f582a57d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 17:04:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ae339cd2e134fd5a545fb8081f41f20\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a28555dcf1448209b31ee94ae9e57d4 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3de8517a7e9d4e1a234336bc35dec6aeb672d40a","revisions":{"3de8517a7e9d4e1a234336bc35dec6aeb672d40a":{"kind":"REWORK","_number":1,"created":"2026-09-25 04:45:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/1007387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.reports","ref":"refs/changes/87/1007387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/87/1007387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/87/1007387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/87/1007387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.reports refs/changes/87/1007387/1"}}},"commit":{"parents":[{"commit":"43a8fac0066ee4acf54ad3a1b0bf14beee68ec26","subject":"Merge \"Ensure recursive detection of package files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/43a8fac0066ee4acf54ad3a1b0bf14beee68ec26"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 04:45:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 04:45:46.000000000","tz":540},"subject":"Declare Python 3.15 support","message":"Declare Python 3.15 support\n\nIt\u0027s now voting in the -next job template.\n\nChange-Id: I747b75e347d70378965ecbb2959f5545960e88a8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/3de8517a7e9d4e1a234336bc35dec6aeb672d40a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/3de8517a7e9d4e1a234336bc35dec6aeb672d40a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
