)]}'
{"id":"openstack%2Foslo.reports~700340","triplet_id":"openstack%2Foslo.reports~master~I5bf3334da567eb480d573ccc1cfe62ae13bdc2a2","project":"openstack/oslo.reports","branch":"master","hashtags":[],"change_id":"I5bf3334da567eb480d573ccc1cfe62ae13bdc2a2","subject":"tox: Trivial cleanup","status":"MERGED","created":"2019-12-22 15:30:09.000000000","updated":"2020-01-28 15:22:35.000000000","submitted":"2020-01-28 15:20:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700340-1580224842639-c94174cc","meta_rev_id":"9a73dd641de0b08cdb50823a1abe2bdfc9ab17dc","_number":700340,"virtual_id_number":700340,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-23 14:08:15.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 15:20:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-01-15 15:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-01-28 13:26:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-01-28 13:26:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 15:21:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-28 13:26:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-28 15:20:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6293102fde32baa898bf54788e7d38d293cd638f","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-22 15:30:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46d6a34774f8f0a5e66ab35d5ce7d9cb8d00a416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 17:03:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/59f5aef63274468e927289cb0ca781b1 : SUCCESS in 1h 28m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e510ab416b344372977011198ef02204 : SUCCESS in 1h 27m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3dea6d00f074334964ed44fbc905051 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca4ee257af1847168280dbeea1c2fa58 : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1358ae3828a14e92843a73cfb186feeb : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fdc822dc3714c3eaea006d3cc796499 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92eea45680cd4c5ea2552db544a8f690 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4530f40a38a42a4b07a47f29d9028c3 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e1f76fbc15b41e2b5b7941eb50fc44b : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1c9c92744d1f85cd13025faa7feb8e271c9c511b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:28:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b7fcb6a4a7c39fbd69162157b7a58a871ea176f","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-23 13:11:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f7c4682216ecfae4ea57fdcc70d4e329672ab60f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 13:34:53.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d4d96fce0caa88f4ab180346d334351a203a774","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-23 13:52:14.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1b9e94874dd9dce6ff63e4e63d5d5a41ee34326b","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-23 14:08:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"deedf4b717569ddd386fe6a36e73dbb7b75dadd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 15:32:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b1276e09376b4dc595048bf6a8c9e94e : SUCCESS in 1h 38m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/419832a08e4f49c5a5d406f78ded68ae : SUCCESS in 1h 29m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1015efdf02b0434eac80b346f3eca678 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f22d66bbd9a94bc0968ac149c747fd48 : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21f9a67766724414a09b28aebd856a65 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8a62290e1c34184aa8de7a7180272d0 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e348f70998494a9c816a50af903d92b4 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7713bfc596d4b5eb3bf004730c20279 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbe8519e8af1455a977798e9e3bc3062 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":3},{"id":"aae57c68e2a68abc053297918ea858b8a2813a1d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-15 15:21:17.000000000","message":"Patch Set 3: Code-Review+2\n\nI meant to +2 and did -2 instead /o\\ oops","accounts_in_message":[],"_revision_number":3},{"id":"d886d5af441ccff6538aedc2767e2bfd6e4bfca9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-28 13:26:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ee4eb8d43d8de43c4605ee9c9d4f407ba83bbaae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 13:26:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4862cd1c8c5b65b4d01ef9dad69a8c8e48f2fbe7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:20:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3e33bcd3a2ad505fb4d0b444dbec64933130f733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:20:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/36f0b7832f20485a8d8cfa318d55f0c7 : SUCCESS in 1h 53m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54962a70595c47629d81c5c43a5d1a4e : SUCCESS in 1h 30m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/028546f0c5b14aa79eec172830dc1b94 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19cbb2246644463cbaba1e3f5b633ddf : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/087533e843094401ad3b454ead84d4ef : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f5c608f1136400cae803adc26230581 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6740ea1994248dcaebe948b3edcb3e4 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ec92057065450b99df80b1f39c6ad1 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44d0cbb02a1746c3875203a64e9ce135 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":3},{"id":"9a73dd641de0b08cdb50823a1abe2bdfc9ab17dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:22:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9234c7bb95f4533aee066bf019ef1dc : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98f1fd7bafcd4c18a8a7999619dab20e : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"edcb44e957d100621617aac6d82de72707f261dc","revisions":{"3c3ea362d77016905d994a8d37f9b68590e1f1d3":{"kind":"REWORK","_number":1,"created":"2019-12-22 15:30:09.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/40/700340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.reports","ref":"refs/changes/40/700340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/1"}}},"commit":{"parents":[{"commit":"1209fb97c72f3634f60e4abde3c703dc93b2843e","subject":"Switch from \u0027collections\u0027 -\u003e \u0027collections.abc\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/1209fb97c72f3634f60e4abde3c703dc93b2843e"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:29:55.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:29:55.000000000","tz":480},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I5bf3334da567eb480d573ccc1cfe62ae13bdc2a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/3c3ea362d77016905d994a8d37f9b68590e1f1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/3c3ea362d77016905d994a8d37f9b68590e1f1d3"}]},"branch":"refs/heads/master"},"fc0b36c52235a3ae1723405721d611a3c9122302":{"kind":"REWORK","_number":2,"created":"2019-12-23 13:11:26.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/40/700340/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.reports","ref":"refs/changes/40/700340/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/2"}}},"commit":{"parents":[{"commit":"1209fb97c72f3634f60e4abde3c703dc93b2843e","subject":"Switch from \u0027collections\u0027 -\u003e \u0027collections.abc\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/1209fb97c72f3634f60e4abde3c703dc93b2843e"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:29:55.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-23 13:11:23.000000000","tz":0},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I5bf3334da567eb480d573ccc1cfe62ae13bdc2a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/fc0b36c52235a3ae1723405721d611a3c9122302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/fc0b36c52235a3ae1723405721d611a3c9122302"}]},"branch":"refs/heads/master"},"edcb44e957d100621617aac6d82de72707f261dc":{"kind":"REWORK","_number":3,"created":"2019-12-23 13:52:14.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/40/700340/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.reports","ref":"refs/changes/40/700340/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.reports refs/changes/40/700340/3"}}},"commit":{"parents":[{"commit":"1209fb97c72f3634f60e4abde3c703dc93b2843e","subject":"Switch from \u0027collections\u0027 -\u003e \u0027collections.abc\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/1209fb97c72f3634f60e4abde3c703dc93b2843e"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:29:55.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-23 13:52:08.000000000","tz":0},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I5bf3334da567eb480d573ccc1cfe62ae13bdc2a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/edcb44e957d100621617aac6d82de72707f261dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/edcb44e957d100621617aac6d82de72707f261dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
