)]}'
{"id":"openstack%2Foslo.reports~777600","triplet_id":"openstack%2Foslo.reports~stable%2Frocky~Id83f06c2f7a120960083b05d01b311e733bd0369","project":"openstack/oslo.reports","branch":"stable/rocky","topic":"cap-bandit-1.6.2","hashtags":[],"change_id":"Id83f06c2f7a120960083b05d01b311e733bd0369","subject":"[stable-only] Cap bandit and fix lower-constraints","status":"MERGED","created":"2021-02-25 14:22:24.000000000","updated":"2021-03-02 11:45:20.000000000","submitted":"2021-03-02 11:43:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777600-cap-bandit-1.6.2","meta_rev_id":"948a9e6d49f9090937b071dc484b2c471d74fc84","_number":777600,"virtual_id_number":777600,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 11:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-03-02 10:15:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-03-01 16:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-03-02 10:15:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 15:50:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 16:00:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-02 10:15:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f6dfea9835fce60b0b0a99c7cf69fb65119d6ec3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-25 14:22:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db3ef48081c44c60833ff5f78d967133157091ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 15:50:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/133096bde82c4930b3c915ecbec6bdbb : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe37b2bbb57f42929de34ac7393855dc : SUCCESS in 5m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/572d367eb7404370875b060b325369c8 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79a4e8480a84d00ae202d18fd94966b : SUCCESS in 5m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3d52395ba9a4e69b1fef45c0aeb5908 : SUCCESS in 3m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3f7c5294ea26490db22cad5d6d740a6e : SUCCESS in 1h 25m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed0409551e8545d89b618d6aedff76ea : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6669f404c68244e09adb02b4a6918035 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f549a40b546df2f094bf774a695b06c6d758d142","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-26 14:11:14.000000000","message":"Patch Set 1:\n\nThis patch fixes py27 stable periodic job failure: http://lists.openstack.org/pipermail/openstack-stable-maint/2021-February/thread.html#30821","accounts_in_message":[],"_revision_number":1},{"id":"e4a1d9d2f660b90181b42be3f168cac435e4f8db","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-01 16:00:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9d7a5ff5553a5e2bb430052ffe238a5e75382ee","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-02 10:15:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ec34e47d8510ed77347f7a7879ed17d59079979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:17:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f846b7526397d91b570efca13a07f1b6fc7eaf03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 11:43:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b296cb77674474ae45568c001ad9b1 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24fc6ec1e7d34ff2bb1c3bcfb855d4ff : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/069b1fb988844fd1b873f8feb0f25bcb : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b733fdd21ca452887f01fe8850f4300 : SUCCESS in 6m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ecc5b525a7ca4683b2b3a0b77c3ab1c1 : SUCCESS in 4m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/303b97d556ff413891fe2df98cfcd41d : SUCCESS in 1h 23m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/945477d52a844f7eb485d68381a0ae0c : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8d5a5c2b09e4e5e9110d2b0a360daef : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"62f624aa0911c4fd2dfc4d46d50000c3a9fea6b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 11:43:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"948a9e6d49f9090937b071dc484b2c471d74fc84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 11:45:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58a43643a184b0aabd624df94ee9d9f : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2644efada5da4b0d80b39ea23880ccb0 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d7d336da64fdea67de9565b95f676996ddaffe6","revisions":{"3d7d336da64fdea67de9565b95f676996ddaffe6":{"kind":"REWORK","_number":1,"created":"2021-02-25 14:22:24.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/00/777600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.reports","ref":"refs/changes/00/777600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/00/777600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/00/777600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.reports refs/changes/00/777600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.reports refs/changes/00/777600/1"}}},"commit":{"parents":[{"commit":"75f1ef441c6629acf34c50903a14d98d1e7ca9e0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/75f1ef441c6629acf34c50903a14d98d1e7ca9e0"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-25 14:21:19.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-25 14:21:19.000000000","tz":60},"subject":"[stable-only] Cap bandit and fix lower-constraints","message":"[stable-only] Cap bandit and fix lower-constraints\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\nSphinx requirements also fixed to make requirements-check job pass.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: Id83f06c2f7a120960083b05d01b311e733bd0369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/3d7d336da64fdea67de9565b95f676996ddaffe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.reports/commit/3d7d336da64fdea67de9565b95f676996ddaffe6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
