)]}'
{"id":"openstack%2Foslo.serialization~418066","triplet_id":"openstack%2Foslo.serialization~master~I4575ea3dad51be9bb2278eb0bfa31cef54c300d5","project":"openstack/oslo.serialization","branch":"master","hashtags":[],"change_id":"I4575ea3dad51be9bb2278eb0bfa31cef54c300d5","subject":"Prefer raising the python2.x type error for b64 decode errors","status":"MERGED","created":"2017-01-09 19:25:14.000000000","updated":"2017-01-23 10:40:32.000000000","submitted":"2017-01-12 01:46:41.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a38421e25a69eaeeb19e9c16f23221ebe0e7543f","_number":418066,"virtual_id_number":418066,"owner":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-01-12 01:46:41.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"date":"2017-01-23 10:40:32.000000000","post_submit":true,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-01-10 12:45:34.000000000","_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":2,"date":"2017-01-12 00:48:10.000000000","_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":1,"date":"2017-01-12 00:48:10.000000000","_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-10 12:45:34.000000000","updated_by":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"reviewer":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"state":"REVIEWER"},{"updated":"2017-01-12 00:48:10.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"},{"updated":"2017-01-12 01:46:41.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-01-23 10:40:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"d67c18ab8c96062585a060fab6a3db8ffc15a1df","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2017-01-09 19:25:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5f3623fea7fd90b582a1bcf2579072070e44664","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-09 20:19:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-oslo.serialization-docs-ubuntu-xenial http://docs-draft.openstack.org/66/418066/1/check/gate-oslo.serialization-docs-ubuntu-xenial/9136076//doc/build/html/ : SUCCESS in 2m 27s\n- gate-oslo.serialization-pep8-ubuntu-xenial http://logs.openstack.org/66/418066/1/check/gate-oslo.serialization-pep8-ubuntu-xenial/1e390e0/ : FAILURE in 2m 39s\n- gate-oslo.serialization-python27-ubuntu-xenial http://logs.openstack.org/66/418066/1/check/gate-oslo.serialization-python27-ubuntu-xenial/ead7f18/ : SUCCESS in 3m 12s\n- gate-oslo.serialization-python35 http://logs.openstack.org/66/418066/1/check/gate-oslo.serialization-python35/46deb4f/ : SUCCESS in 4m 05s\n- gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial http://logs.openstack.org/66/418066/1/check/gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial/5546591/ : SUCCESS in 53m 37s\n- gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial-newton http://logs.openstack.org/66/418066/1/check/gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial-newton/00e3e83/ : FAILURE in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"718844b274c5ea30129fb89f94261412f0165318","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2017-01-10 00:37:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5da588d2a26ea8acf6fd2191988e8e93b60bd72e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-10 02:16:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-oslo.serialization-docs-ubuntu-xenial http://docs-draft.openstack.org/66/418066/2/check/gate-oslo.serialization-docs-ubuntu-xenial/02d8847//doc/build/html/ : SUCCESS in 2m 45s\n- gate-oslo.serialization-pep8-ubuntu-xenial http://logs.openstack.org/66/418066/2/check/gate-oslo.serialization-pep8-ubuntu-xenial/a9b9840/ : SUCCESS in 3m 58s\n- gate-oslo.serialization-python27-ubuntu-xenial http://logs.openstack.org/66/418066/2/check/gate-oslo.serialization-python27-ubuntu-xenial/224bc63/ : SUCCESS in 1m 43s\n- gate-oslo.serialization-python35 http://logs.openstack.org/66/418066/2/check/gate-oslo.serialization-python35/2ce0e88/ : SUCCESS in 3m 10s\n- gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial http://logs.openstack.org/66/418066/2/check/gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial/2f53542/ : FAILURE in 1h 38m 50s\n- gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial-newton http://logs.openstack.org/66/418066/2/check/gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial-newton/5022bf1/ : FAILURE in 7m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a785c81f02796a0f24faad04cb807fe9ac27def","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-01-10 12:45:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"337188ce6cd6e573f0ece8725bf612a8ad1d4895","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2017-01-10 19:50:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"22cbef56c5a700f11559af69a75c251280ac942c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-10 20:56:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.serialization-docs-ubuntu-xenial http://docs-draft.openstack.org/66/418066/2/check/gate-oslo.serialization-docs-ubuntu-xenial/8791e21//doc/build/html/ : SUCCESS in 2m 01s\n- gate-oslo.serialization-pep8-ubuntu-xenial http://logs.openstack.org/66/418066/2/check/gate-oslo.serialization-pep8-ubuntu-xenial/5a19250/ : SUCCESS in 2m 01s\n- gate-oslo.serialization-python27-ubuntu-xenial http://logs.openstack.org/66/418066/2/check/gate-oslo.serialization-python27-ubuntu-xenial/d114821/ : SUCCESS in 1m 48s\n- gate-oslo.serialization-python35 http://logs.openstack.org/66/418066/2/check/gate-oslo.serialization-python35/1284e36/ : SUCCESS in 2m 01s\n- gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial http://logs.openstack.org/66/418066/2/check/gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial/9552367/ : SUCCESS in 1h 04m 02s\n- gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial-newton http://logs.openstack.org/66/418066/2/check/gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial-newton/7ea1279/ : FAILURE in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfb0c4d5d5101feb946025f6cef498a491f33bde","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-01-12 00:48:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"297746a3bbca651a810c011f4f7ff4e34a7c4e89","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-12 00:48:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"6361e291a41f5b0dd3a7a86f7a6527ff517ddd13","date":"2017-01-12 01:46:41.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"94aed2be7396d21ac7eb9fc17a0dd377b56f5b42","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-12 01:46:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-oslo.serialization-docs-ubuntu-xenial http://docs-draft.openstack.org/66/418066/2/gate/gate-oslo.serialization-docs-ubuntu-xenial/ee18d93//doc/build/html/ : SUCCESS in 2m 32s\n- gate-oslo.serialization-pep8-ubuntu-xenial http://logs.openstack.org/66/418066/2/gate/gate-oslo.serialization-pep8-ubuntu-xenial/0342858/ : SUCCESS in 3m 12s\n- gate-oslo.serialization-python27-ubuntu-xenial http://logs.openstack.org/66/418066/2/gate/gate-oslo.serialization-python27-ubuntu-xenial/7f26994/ : SUCCESS in 2m 24s\n- gate-oslo.serialization-python35 http://logs.openstack.org/66/418066/2/gate/gate-oslo.serialization-python35/1595be0/ : SUCCESS in 1m 50s\n- gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial http://logs.openstack.org/66/418066/2/gate/gate-tempest-dsvm-neutron-src-oslo.serialization-ubuntu-xenial/af13498/ : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a38421e25a69eaeeb19e9c16f23221ebe0e7543f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-01-23 10:40:32.000000000","message":"Patch Set 2:\n\nWhoops - sorry about that. Don\u0027t know why I didn\u0027t prefer the (more widely used) Python 2 exception variant, as done here.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac5787d0e42ba6aa1c26d2e040c9f0a1ca8cb575","revisions":{"f011b19686244fe1137296d78dd807f5265a9b82":{"kind":"REWORK","_number":1,"created":"2017-01-09 19:25:14.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/66/418066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.serialization","ref":"refs/changes/66/418066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.serialization refs/changes/66/418066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.serialization refs/changes/66/418066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.serialization refs/changes/66/418066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.serialization refs/changes/66/418066/1"}}},"commit":{"parents":[{"commit":"004bc475222f57f6840157be9fe3dd8325832b1e","subject":"Merge \"Replace six.iteritems() with .items()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.serialization/commit/004bc475222f57f6840157be9fe3dd8325832b1e"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2017-01-09 19:23:40.000000000","tz":-480},"committer":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2017-01-09 19:25:08.000000000","tz":-480},"subject":"Prefer raising the python2.x type error for b64 decode errors","message":"Prefer raising the python2.x type error for b64 decode errors\n\nThe current change made in de68f08d37 breaks cinder and other\nprojects that were expecting a TypeError, so for now and to keep\nthose projects operating translate the py3.x exception from binascii\ninto a type error to prefer consistency with existing code.\n\nChange-Id: I4575ea3dad51be9bb2278eb0bfa31cef54c300d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.serialization/commit/f011b19686244fe1137296d78dd807f5265a9b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.serialization/commit/f011b19686244fe1137296d78dd807f5265a9b82"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"004bc475222f57f6840157be9fe3dd8325832b1e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ac5787d0e42ba6aa1c26d2e040c9f0a1ca8cb575":{"kind":"REWORK","_number":2,"created":"2017-01-10 00:37:32.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/66/418066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.serialization","ref":"refs/changes/66/418066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.serialization refs/changes/66/418066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.serialization refs/changes/66/418066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.serialization refs/changes/66/418066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.serialization refs/changes/66/418066/2"}}},"commit":{"parents":[{"commit":"004bc475222f57f6840157be9fe3dd8325832b1e","subject":"Merge \"Replace six.iteritems() with .items()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.serialization/commit/004bc475222f57f6840157be9fe3dd8325832b1e"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2017-01-09 19:23:40.000000000","tz":-480},"committer":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2017-01-10 00:37:28.000000000","tz":-480},"subject":"Prefer raising the python2.x type error for b64 decode errors","message":"Prefer raising the python2.x type error for b64 decode errors\n\nThe current change made in de68f08d37 breaks cinder and other\nprojects that were expecting a TypeError, so for now and to keep\nthose projects operating translate the py3.x exception from binascii\ninto a type error to prefer consistency with existing code.\n\nChange-Id: I4575ea3dad51be9bb2278eb0bfa31cef54c300d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.serialization/commit/ac5787d0e42ba6aa1c26d2e040c9f0a1ca8cb575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.serialization/commit/ac5787d0e42ba6aa1c26d2e040c9f0a1ca8cb575"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"004bc475222f57f6840157be9fe3dd8325832b1e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
