)]}'
{"id":"openstack%2Foslo.service~658249","triplet_id":"openstack%2Foslo.service~master~Ie3dd6c91c733b85c233213a1c6343009d5f34766","project":"openstack/oslo.service","branch":"master","topic":"fix_bandit_error","hashtags":[],"change_id":"Ie3dd6c91c733b85c233213a1c6343009d5f34766","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","status":"MERGED","created":"2019-05-10 03:37:40.000000000","updated":"2019-05-14 19:08:51.000000000","submitted":"2019-05-14 19:08:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658249-1557860931668-96febb6b","meta_rev_id":"c9175e97f51195b8b67522446bd78abbf7472f48","_number":658249,"virtual_id_number":658249,"owner":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"date":"2019-05-14 15:23:11.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-05-14 15:10:31.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"date":"2019-05-14 12:55:00.000000000","_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":2,"date":"2019-05-14 19:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-14 14:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-05-14 15:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-05-14 15:16:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 05:35:56.000000000","updated_by":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"reviewer":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"state":"REVIEWER"},{"updated":"2019-05-10 14:51:18.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-05-14 14:39:16.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-14 15:10:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-14 15:16:41.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-05-14 15:23:11.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-14 19:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2dfd9b2753fc61143454dbe0e85f30c5ac4ac7ad","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-10 03:37:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9742063814f367ab8f2f2d6f9a0c322546df64c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 05:30:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/49/658249/1/check/requirements-check/2ed04d2/ : SUCCESS in 3m 21s\n- tempest-full http://logs.openstack.org/49/658249/1/check/tempest-full/10555bd/ : SUCCESS in 1h 51m 12s\n- tempest-full-py3 http://logs.openstack.org/49/658249/1/check/tempest-full-py3/341b29f/ : SUCCESS in 1h 28m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/658249/1/check/openstack-tox-lower-constraints/f7fe558/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/49/658249/1/check/openstack-tox-pep8/5ebba2b/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/49/658249/1/check/openstack-tox-py27/d1d5dc8/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/49/658249/1/check/openstack-tox-py35/7319e34/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/49/658249/1/check/openstack-tox-py36/8ebf8ca/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/49/658249/1/check/openstack-tox-py37/9b7cf84/ : SUCCESS in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/49/658249/1/check/openstack-tox-docs/7ab71a3/html/ : SUCCESS in 5m 32s\n- build-openstack-releasenotes http://logs.openstack.org/49/658249/1/check/build-openstack-releasenotes/8ff834f/html/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"371d7d6f3999ffd0df18d6cc2d5932e93a159a6c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-10 08:25:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"13984cba128ff969113c79dcd64e321f09b26f60","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-10 14:51:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f835ca59eac1295cc3d6efc0bf49c1e8f7abfce2","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-10 15:40:43.000000000","message":"Patch Set 1: Code-Review+1\n\nThis broke a lot of people: https://github.com/PyCQA/bandit/issues/488\n\nI imagine it\u0027s going to be fixed very soon, so another option is to just require bandit!\u003d1.60. Other projects are taking that path: https://github.com/openstack/python-openstackclient/commit/6385d64237c9973dd4c7dd53efb6664ea2c719da","accounts_in_message":[],"_revision_number":1},{"id":"27274aaacbb04cdff9f5a819d2e3040ba6a7ee99","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-11 02:44:29.000000000","message":"Patch Set 1:\n\n\u003e This broke a lot of people: https://github.com/PyCQA/bandit/issues/488\n \u003e \n \u003e I imagine it\u0027s going to be fixed very soon, so another option is to\n \u003e just require bandit!\u003d1.60. Other projects are taking that path:\n \u003e https://github.com/openstack/python-openstackclient/commit/6385d64237c9973dd4c7dd53efb6664ea2c719da\n\nYes, that\u0027s a good idea, thanks. I will update this commit.","accounts_in_message":[],"_revision_number":1},{"id":"1f86643054cd298cd7762432e65e58a43e254e1e","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-11 02:53:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"628ea258ee593ca6a1f0618e7d21fc094571a275","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 04:28:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/49/658249/2/check/requirements-check/9711c2c/ : SUCCESS in 3m 19s\n- tempest-full http://logs.openstack.org/49/658249/2/check/tempest-full/b2e7b87/ : SUCCESS in 1h 26m 33s\n- tempest-full-py3 http://logs.openstack.org/49/658249/2/check/tempest-full-py3/a8d142b/ : SUCCESS in 1h 29m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/658249/2/check/openstack-tox-lower-constraints/41581d5/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/49/658249/2/check/openstack-tox-pep8/f908102/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/49/658249/2/check/openstack-tox-py27/c42fd39/ : SUCCESS in 4m 45s\n- openstack-tox-py35 http://logs.openstack.org/49/658249/2/check/openstack-tox-py35/74a4bd0/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/49/658249/2/check/openstack-tox-py36/432207d/ : SUCCESS in 5m 35s\n- openstack-tox-py37 http://logs.openstack.org/49/658249/2/check/openstack-tox-py37/919a4da/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/49/658249/2/check/openstack-tox-docs/3833ef2/html/ : SUCCESS in 3m 58s\n- build-openstack-releasenotes http://logs.openstack.org/49/658249/2/check/build-openstack-releasenotes/fb85c70/html/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"69ddd65ede439623f4ebdca182ec42c23fbc344c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-13 10:00:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3c3d8abf44caf9ca807f0f0eb102dae7734a6ff2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 11:59:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/658249/3/check/requirements-check/c0a6f2a/ : FAILURE in 3m 29s\n- tempest-full http://logs.openstack.org/49/658249/3/check/tempest-full/5e33289/ : SUCCESS in 1h 57m 22s\n- tempest-full-py3 http://logs.openstack.org/49/658249/3/check/tempest-full-py3/3257bde/ : SUCCESS in 1h 41m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/658249/3/check/openstack-tox-lower-constraints/4fa265e/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/49/658249/3/check/openstack-tox-pep8/94af37f/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/49/658249/3/check/openstack-tox-py27/44348a3/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/49/658249/3/check/openstack-tox-py35/b4bf490/ : SUCCESS in 6m 19s\n- openstack-tox-py36 http://logs.openstack.org/49/658249/3/check/openstack-tox-py36/584ce61/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/49/658249/3/check/openstack-tox-py37/4c1255a/ : SUCCESS in 9m 02s\n- openstack-tox-docs http://logs.openstack.org/49/658249/3/check/openstack-tox-docs/244c604/html/ : SUCCESS in 4m 19s\n- build-openstack-releasenotes http://logs.openstack.org/49/658249/3/check/build-openstack-releasenotes/e1ce2c5/html/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3},{"id":"d43fdddbe9b627888f5b833ccf06ae076b8e78d7","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-13 12:53:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4b2a64e6df7738f48f4949a3b2e8670180e04163","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-13 13:35:25.000000000","message":"Patch Set 3:\n\n\u003e recheck\n\nI guess your CI problem is due to sphinx.\n\nSee this one \u003d\u003e https://review.opendev.org/658812","accounts_in_message":[],"_revision_number":3},{"id":"f46cc070e21f54ac7b31b3efa4099d6136392209","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-13 13:42:49.000000000","message":"Patch Set 3:\n\nI\u0027ve submit a patch to fix it https://review.opendev.org/#/c/658818/","accounts_in_message":[],"_revision_number":3},{"id":"5ce747d6a89cc470a8f76cfa73773c854822426b","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-13 14:13:27.000000000","message":"Patch Set 3:\n\nAfter Hervé Beraud\u0027s commit: \"Fix Sphinx lower-constraints\" https://review.opendev.org/#/c/658818/ is merged, a recheck command is needed here.","accounts_in_message":[],"_revision_number":3},{"id":"09e18dc282ee5400473621efed3282b5c5def8bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 14:29:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/658249/3/check/requirements-check/81efbd5/ : FAILURE in 3m 17s\n- tempest-full http://logs.openstack.org/49/658249/3/check/tempest-full/98363e3/ : FAILURE in 25m 55s\n- tempest-full-py3 http://logs.openstack.org/49/658249/3/check/tempest-full-py3/1bafec2/ : SUCCESS in 1h 31m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/658249/3/check/openstack-tox-lower-constraints/07fafc1/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/49/658249/3/check/openstack-tox-pep8/5cf6e21/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/49/658249/3/check/openstack-tox-py27/d279021/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/49/658249/3/check/openstack-tox-py35/d47c476/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/49/658249/3/check/openstack-tox-py36/5e3416e/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/49/658249/3/check/openstack-tox-py37/a598c77/ : SUCCESS in 5m 28s\n- openstack-tox-docs http://logs.openstack.org/49/658249/3/check/openstack-tox-docs/9dcad70/html/ : SUCCESS in 4m 22s\n- build-openstack-releasenotes http://logs.openstack.org/49/658249/3/check/build-openstack-releasenotes/3fde8ec/html/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"73839816112687d5230dbde1fafe770a62794452","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-13 20:06:28.000000000","message":"Patch Set 3: Code-Review-1\n\nPlease see https://review.opendev.org/#/c/658867/ for an example of how we decided to handle this. Note that you will need both the bandit and sphinx changes, and please add the followup patch to uncap bandit so we\u0027re ready to go when bandit releases a fixed version. Thanks.","accounts_in_message":[],"_revision_number":3},{"id":"7e93b99c02a618ee46bc8720cbe4cfb08f216891","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 03:35:03.000000000","message":"Patch Set 3:\n\n\u003e Please see https://review.opendev.org/#/c/658867/ for an example of\n \u003e how we decided to handle this. Note that you will need both the\n \u003e bandit and sphinx changes, and please add the followup patch to\n \u003e uncap bandit so we\u0027re ready to go when bandit releases a fixed\n \u003e version. Thanks.\n\nOk.","accounts_in_message":[],"_revision_number":3},{"id":"f47c711f2b86d0052c91239818915a172261cf62","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 05:28:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6000484145fef3a76507056a0f37b39b83552fba","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 05:58:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7539f33a701be0590ba45ae42b2dacd3b0d7eaa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 07:38:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/658249/5/check/requirements-check/7aba6f7/ : FAILURE in 3m 43s\n- tempest-full http://logs.openstack.org/49/658249/5/check/tempest-full/88446d2/ : SUCCESS in 1h 38m 41s\n- tempest-full-py3 http://logs.openstack.org/49/658249/5/check/tempest-full-py3/ee84c9a/ : SUCCESS in 1h 20m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/658249/5/check/openstack-tox-lower-constraints/6339928/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/49/658249/5/check/openstack-tox-pep8/bf01143/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/49/658249/5/check/openstack-tox-py27/2f83253/ : SUCCESS in 5m 04s\n- openstack-tox-py35 http://logs.openstack.org/49/658249/5/check/openstack-tox-py35/4c946bf/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/49/658249/5/check/openstack-tox-py36/fa17055/ : SUCCESS in 5m 13s\n- openstack-tox-py37 http://logs.openstack.org/49/658249/5/check/openstack-tox-py37/b6396ff/ : SUCCESS in 8m 45s\n- openstack-tox-docs http://logs.openstack.org/49/658249/5/check/openstack-tox-docs/89b5181/html/ : SUCCESS in 4m 35s\n- build-openstack-releasenotes http://logs.openstack.org/49/658249/5/check/build-openstack-releasenotes/af51de2/html/ : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":5},{"id":"1c7ea369b3eaa38cfba50c8a6e5669dc2ce1706b","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 07:54:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"69c325042d6ebbfc020042827ccf6082b27d1149","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 08:27:00.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nFYI I\u0027ve close mine to simplify reviews on this topic and the doomsday dependencies :)","accounts_in_message":[],"_revision_number":6},{"id":"95f6129a53b95c0e6ddd1080a747e520d0314415","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 08:43:16.000000000","message":"Patch Set 6:\n\n\u003e (2 comments)\n \u003e \n \u003e FYI I\u0027ve close mine to simplify reviews on this topic and the\n \u003e doomsday dependencies :)\n\nPlease look this: http://logs.openstack.org/49/658249/6/check/openstack-tox-pep8/bdbc7d6/job-output.txt.gz","accounts_in_message":[],"_revision_number":6},{"id":"76eb192e9133de8e1a75d873ad5d5bd979b52b51","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 08:48:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"22b33b337b55ecb16a33f1e17c88621d37cf4a2a","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 08:51:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4716f1b53bf3d60c5ac33eaa01ad8d67ec2e6cc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 10:20:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/658249/7/check/requirements-check/4065967/ : FAILURE in 3m 41s\n- tempest-full http://logs.openstack.org/49/658249/7/check/tempest-full/0f88a75/ : FAILURE in 41m 28s\n- tempest-full-py3 http://logs.openstack.org/49/658249/7/check/tempest-full-py3/012fb77/ : SUCCESS in 1h 27m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/658249/7/check/openstack-tox-lower-constraints/b185306/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/49/658249/7/check/openstack-tox-pep8/5125571/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/49/658249/7/check/openstack-tox-py27/3750540/ : SUCCESS in 4m 21s\n- openstack-tox-py35 http://logs.openstack.org/49/658249/7/check/openstack-tox-py35/168dd77/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/49/658249/7/check/openstack-tox-py36/5e985db/ : SUCCESS in 4m 38s\n- openstack-tox-py37 http://logs.openstack.org/49/658249/7/check/openstack-tox-py37/a989d0e/ : SUCCESS in 4m 42s\n- openstack-tox-docs http://logs.openstack.org/49/658249/7/check/openstack-tox-docs/0e31791/html/ : SUCCESS in 4m 53s\n- build-openstack-releasenotes http://logs.openstack.org/49/658249/7/check/build-openstack-releasenotes/0464269/html/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":7},{"id":"3afcfce31907b9a3edcd27be435ea7cde75c9901","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 12:17:11.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7cf6716d861cc206ff075f3093fcbb03baeec7de","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 12:18:37.000000000","message":"Patch Set 7:\n\nMy required changes are related to:\nhttp://logs.openstack.org/49/658249/7/check/requirements-check/4065967/job-output.txt.gz#_2019-05-14_08_55_04_311813","accounts_in_message":[],"_revision_number":7},{"id":"3169a53f45746ed91dc92332b45005bfcd43645e","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 12:27:23.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"397048385b7bede05ba33862ebfd617b36c3be1d","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 12:28:45.000000000","message":"Patch Set 8:\n\n\u003e (1 comment)\n\nThank you.","accounts_in_message":[],"_revision_number":8},{"id":"6777c28d2d7da630b7e1b87599b5fa8abfeca819","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 12:48:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7f03dc9f22070321c9d8847b352fc167ca8d376d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-14 12:51:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"aa324756b8e61607f359d91330ddd8ac382965bc","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-14 12:55:00.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"398d30b7df85bc5a4e7eb0144f93a013909299fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 13:04:16.000000000","message":"Patch Set 9:\n\n\u003e (1 comment)\n\nYeah sorry I wanted to say to you to remove that (lower-constraints changes). I think it\u0027s now ok.\n\nWait for the CI","accounts_in_message":[],"_revision_number":9},{"id":"03bbb1d077d241372ebb1ab0ceee425b5c7fb5a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 14:23:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/658249/9/check/requirements-check/f19fc49/ : SUCCESS in 3m 56s\n- tempest-full http://logs.openstack.org/49/658249/9/check/tempest-full/6f27987/ : SUCCESS in 1h 32m 22s\n- tempest-full-py3 http://logs.openstack.org/49/658249/9/check/tempest-full-py3/38dfd16/ : FAILURE in 1h 14m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/658249/9/check/openstack-tox-lower-constraints/b5a0df7/ : SUCCESS in 6m 15s\n- openstack-tox-pep8 http://logs.openstack.org/49/658249/9/check/openstack-tox-pep8/cd607eb/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/49/658249/9/check/openstack-tox-py27/3630acd/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/49/658249/9/check/openstack-tox-py35/fff2147/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/49/658249/9/check/openstack-tox-py36/ef44b4c/ : SUCCESS in 6m 21s\n- openstack-tox-py37 http://logs.openstack.org/49/658249/9/check/openstack-tox-py37/aa893ba/ : SUCCESS in 6m 40s\n- openstack-tox-docs http://logs.openstack.org/49/658249/9/check/openstack-tox-docs/8463e88/html/ : SUCCESS in 4m 57s\n- build-openstack-releasenotes http://logs.openstack.org/49/658249/9/check/build-openstack-releasenotes/c6ac9c4/html/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":9},{"id":"004f589f3e8c52a38aef1a426c6e1059d90020dd","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-14 14:39:16.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)\n\nrecheck\n\nCan you also push an uncap followup patch? Something like https://review.opendev.org/#/c/658911/","accounts_in_message":[],"_revision_number":9},{"id":"36c87372da655e0e89282ca0647f730c28420297","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 15:07:42.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"24cb96e7325f4e6f6e49dbffb3c660e3028b791e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 15:10:31.000000000","message":"Patch Set 9:\n\n\u003e (1 comment)\n \u003e \n \u003e recheck\n \u003e \n \u003e Can you also push an uncap followup patch? Something like\n \u003e https://review.opendev.org/#/c/658911/\n\nUncap patch \u003d\u003e https://review.opendev.org/#/c/659095/","accounts_in_message":[],"_revision_number":9},{"id":"a86f37065b397ca365af03a7bd25a0ba44291f3c","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-14 15:16:41.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"3998f2af45684e34969f91eb69b47a84d05a4d3a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-14 15:23:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5aab412be333544e3c680025a560239965ba5742","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-14 15:31:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0f85ea05cbad317cb81b40c5b6f4b2f5dba49ec5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 17:06:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/49/658249/9/check/requirements-check/948d209/ : SUCCESS in 3m 16s\n- tempest-full http://logs.openstack.org/49/658249/9/check/tempest-full/19ff75e/ : SUCCESS in 1h 50m 35s\n- tempest-full-py3 http://logs.openstack.org/49/658249/9/check/tempest-full-py3/93a5815/ : SUCCESS in 1h 46m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/658249/9/check/openstack-tox-lower-constraints/2b7d678/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/49/658249/9/check/openstack-tox-pep8/4b9e580/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/49/658249/9/check/openstack-tox-py27/a0b22b7/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/49/658249/9/check/openstack-tox-py35/c14723c/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/49/658249/9/check/openstack-tox-py36/bc5886c/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/49/658249/9/check/openstack-tox-py37/9e5f555/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/49/658249/9/check/openstack-tox-docs/efca485/html/ : SUCCESS in 4m 07s\n- build-openstack-releasenotes http://logs.openstack.org/49/658249/9/check/build-openstack-releasenotes/7c83802/html/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":9},{"id":"353a00d06882f3830d1b8b3e59020183287d3f65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 17:06:46.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4e091f2b2aa6fe46e59da4ef88d29eabd38033b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 19:08:51.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/49/658249/9/gate/requirements-check/b604513/ : SUCCESS in 5m 03s\n- tempest-full http://logs.openstack.org/49/658249/9/gate/tempest-full/e9a467e/ : SUCCESS in 1h 59m 00s\n- tempest-full-py3 http://logs.openstack.org/49/658249/9/gate/tempest-full-py3/e639658/ : SUCCESS in 1h 43m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/658249/9/gate/openstack-tox-lower-constraints/deeb978/ : SUCCESS in 5m 41s\n- openstack-tox-pep8 http://logs.openstack.org/49/658249/9/gate/openstack-tox-pep8/323748b/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/49/658249/9/gate/openstack-tox-py27/579201c/ : SUCCESS in 6m 07s\n- openstack-tox-py35 http://logs.openstack.org/49/658249/9/gate/openstack-tox-py35/14b2d2a/ : SUCCESS in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/49/658249/9/gate/openstack-tox-py36/20dec4f/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/49/658249/9/gate/openstack-tox-py37/dd4263d/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/49/658249/9/gate/openstack-tox-docs/4a42a4b/html/ : SUCCESS in 5m 32s\n- build-openstack-releasenotes http://logs.openstack.org/49/658249/9/gate/build-openstack-releasenotes/38b4a37/html/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":9},{"id":"05ce70bf58c425924c3a35212e44d5d3fa5e30c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 19:08:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"6aa2b7159c38bf19129335d6856b264fa9dffdfa","revisions":{"26d4b77c7e633c18444fca6191deaeaca4b567fa":{"kind":"REWORK","_number":1,"created":"2019-05-10 03:37:40.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/49/658249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/49/658249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/1"}}},"commit":{"parents":[{"commit":"99eb0bc18e8e803a832a59ee0552e52593590a6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/99eb0bc18e8e803a832a59ee0552e52593590a6c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"subject":"Fix Pep8 bandit error","message":"Fix Pep8 bandit error\n\nMake bandit runs with 1.6.0\n\nbandit 1.6.0 was released today:\nhttps://pypi.org/project/bandit/1.6.0/\n\nAnd is uncapped in test-requirements:\nhttps://github.com/openstack/oslo.service/blob/master/test-requirements.txt#L21\n\nAnd not in upper-constraints:\nhttps://github.com/openstack/requirements/blob/master/upper-constraints.txt\n\nThe -x option for bandit changed in 1.6.0 and now\nsupports glob patterns so use that to correctly\nexclude test code from bandit scans.\n\nSeems we should exclude test code from bandit scans.\n\nChange-Id: Ie3dd6c91c733b85c233213a1c6343009d5f34766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/26d4b77c7e633c18444fca6191deaeaca4b567fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/26d4b77c7e633c18444fca6191deaeaca4b567fa"}]},"branch":"refs/heads/master"},"676eed7f9fd7ccbd905c744a2a0ae9bcb8753811":{"kind":"REWORK","_number":2,"created":"2019-05-11 02:53:38.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/49/658249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/49/658249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/2"}}},"commit":{"parents":[{"commit":"99eb0bc18e8e803a832a59ee0552e52593590a6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/99eb0bc18e8e803a832a59ee0552e52593590a6c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-11 02:53:05.000000000","tz":480},"subject":"Fix Pep8 bandit error","message":"Fix Pep8 bandit error\n\nBlacklist Bandit 1.6.0 due to directory exclusion bug\n\nBandit 1.6.0 introduces a regression[0] with the -x option, a fix\nis expected to be included in 1.6.1 soon.\n\n[0] PyCQA/bandit#488\n[1] PyCQA/bandit#489\n\nChange-Id: Ie3dd6c91c733b85c233213a1c6343009d5f34766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/676eed7f9fd7ccbd905c744a2a0ae9bcb8753811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/676eed7f9fd7ccbd905c744a2a0ae9bcb8753811"}]},"branch":"refs/heads/master"},"32d3f838059edab0c2b12a5f43b537fb79f94b16":{"kind":"REWORK","_number":3,"created":"2019-05-13 10:00:28.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/49/658249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/49/658249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/3"}}},"commit":{"parents":[{"commit":"99eb0bc18e8e803a832a59ee0552e52593590a6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/99eb0bc18e8e803a832a59ee0552e52593590a6c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-13 09:58:35.000000000","tz":480},"subject":"Fix Pep8 bandit error","message":"Fix Pep8 bandit error\n\nMake bandit runs with 1.6.0\n\nThe -x option for bandit changed in 1.6.0 and now\nsupports glob patterns so use that to correctly\nexclude test code from bandit scans.\n\nChange-Id: Ie3dd6c91c733b85c233213a1c6343009d5f34766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/32d3f838059edab0c2b12a5f43b537fb79f94b16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/32d3f838059edab0c2b12a5f43b537fb79f94b16"}]},"branch":"refs/heads/master"},"60f1e29c2848ea44ca67d6f3a3756bdda8e2830f":{"kind":"REWORK","_number":4,"created":"2019-05-14 05:28:51.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/49/658249/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/49/658249/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/4"}}},"commit":{"parents":[{"commit":"99eb0bc18e8e803a832a59ee0552e52593590a6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/99eb0bc18e8e803a832a59ee0552e52593590a6c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-14 05:27:14.000000000","tz":480},"subject":"Fix Pep8 bandit error and update Sphinx requirement","message":"Fix Pep8 bandit error and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start capping\nit there as well.\n\nReference: https://github.com/PyCQA/bandit/pull/489\n\nChange-Id: Ie3dd6c91c733b85c233213a1c6343009d5f34766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/60f1e29c2848ea44ca67d6f3a3756bdda8e2830f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/60f1e29c2848ea44ca67d6f3a3756bdda8e2830f"}]},"branch":"refs/heads/master"},"39839e8226acf604f8b4115057a15beecffa5b40":{"kind":"REWORK","_number":5,"created":"2019-05-14 05:58:51.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/49/658249/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/49/658249/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/5"}}},"commit":{"parents":[{"commit":"99eb0bc18e8e803a832a59ee0552e52593590a6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/99eb0bc18e8e803a832a59ee0552e52593590a6c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-14 05:50:26.000000000","tz":480},"subject":"Fix Pep8 bandit error and update Sphinx lower-constraints","message":"Fix Pep8 bandit error and update Sphinx lower-constraints\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start marking\nit there as well.\n\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\nCo-Authored-By: Ben Nemec \u003copenstack@nemebean.com\u003e\n\nReference:\nhttps://github.com/PyCQA/bandit/pull/489\n\nRelated to Ibaa10b092b262e49435d35f093906bb75dc9762f\n\nChange-Id: Ie3dd6c91c733b85c233213a1c6343009d5f34766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/39839e8226acf604f8b4115057a15beecffa5b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/39839e8226acf604f8b4115057a15beecffa5b40"}]},"branch":"refs/heads/master"},"ca1b4ac2e072e7f7b7950ac84b9c6dea388893c4":{"kind":"REWORK","_number":6,"created":"2019-05-14 07:54:54.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/49/658249/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/49/658249/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/6"}}},"commit":{"parents":[{"commit":"99eb0bc18e8e803a832a59ee0552e52593590a6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/99eb0bc18e8e803a832a59ee0552e52593590a6c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-14 07:54:22.000000000","tz":480},"subject":"Fix Pep8 bandit error and update Sphinx lower-constraints","message":"Fix Pep8 bandit error and update Sphinx lower-constraints\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start marking\nit there as well.\n\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\nCo-Authored-By: Ben Nemec \u003copenstack@nemebean.com\u003e\n\nReference:\nhttps://github.com/PyCQA/bandit/pull/489\n\nRelated to Ibaa10b092b262e49435d35f093906bb75dc9762f\n\nChange-Id: Ie3dd6c91c733b85c233213a1c6343009d5f34766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/ca1b4ac2e072e7f7b7950ac84b9c6dea388893c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/ca1b4ac2e072e7f7b7950ac84b9c6dea388893c4"}]},"branch":"refs/heads/master"},"768de2c0f3bdf07596b158288ed74a5f1acbcee2":{"kind":"REWORK","_number":7,"created":"2019-05-14 08:51:42.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/49/658249/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/49/658249/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/7"}}},"commit":{"parents":[{"commit":"99eb0bc18e8e803a832a59ee0552e52593590a6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/99eb0bc18e8e803a832a59ee0552e52593590a6c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-14 08:50:34.000000000","tz":480},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start marking\nit there as well.\n\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\nCo-Authored-By: Ben Nemec \u003copenstack@nemebean.com\u003e\n\nReference:\nhttps://github.com/PyCQA/bandit/pull/489\n\nRelated to Ibaa10b092b262e49435d35f093906bb75dc9762f\n\nChange-Id: Ie3dd6c91c733b85c233213a1c6343009d5f34766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/768de2c0f3bdf07596b158288ed74a5f1acbcee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/768de2c0f3bdf07596b158288ed74a5f1acbcee2"}]},"branch":"refs/heads/master"},"ae95580abb983a3c6437ce6c04e1ab5e568200de":{"kind":"REWORK","_number":8,"created":"2019-05-14 12:27:23.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/49/658249/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/49/658249/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/8"}}},"commit":{"parents":[{"commit":"99eb0bc18e8e803a832a59ee0552e52593590a6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/99eb0bc18e8e803a832a59ee0552e52593590a6c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-14 12:26:32.000000000","tz":480},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start marking\nit there as well.\n\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\nCo-Authored-By: Ben Nemec \u003copenstack@nemebean.com\u003e\n\nReference:\nhttps://github.com/PyCQA/bandit/pull/489\n\nRelated to Ibaa10b092b262e49435d35f093906bb75dc9762f\n\nChange-Id: Ie3dd6c91c733b85c233213a1c6343009d5f34766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/ae95580abb983a3c6437ce6c04e1ab5e568200de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/ae95580abb983a3c6437ce6c04e1ab5e568200de"}]},"branch":"refs/heads/master"},"6aa2b7159c38bf19129335d6856b264fa9dffdfa":{"kind":"REWORK","_number":9,"created":"2019-05-14 12:48:34.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/49/658249/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/49/658249/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/49/658249/9"}}},"commit":{"parents":[{"commit":"99eb0bc18e8e803a832a59ee0552e52593590a6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/99eb0bc18e8e803a832a59ee0552e52593590a6c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-10 03:31:00.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-05-14 12:46:18.000000000","tz":480},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start marking\nit there as well.\n\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\nCo-Authored-By: Ben Nemec \u003copenstack@nemebean.com\u003e\n\nReference:\nhttps://github.com/PyCQA/bandit/pull/489\n\nRelated to Ibaa10b092b262e49435d35f093906bb75dc9762f\n\nChange-Id: Ie3dd6c91c733b85c233213a1c6343009d5f34766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/6aa2b7159c38bf19129335d6856b264fa9dffdfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/6aa2b7159c38bf19129335d6856b264fa9dffdfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
