)]}'
{"id":"openstack%2Foslo.service~658992","triplet_id":"openstack%2Foslo.service~master~I3f86f4867eb0cd5010abadf68620aa3450d3e64d","project":"openstack/oslo.service","branch":"master","topic":"make-pid-accessible-to-backdoor-socket-path","hashtags":[],"change_id":"I3f86f4867eb0cd5010abadf68620aa3450d3e64d","subject":"Make PID availabe as formatstring in backdoor path","status":"MERGED","created":"2019-05-14 09:09:59.000000000","updated":"2019-05-30 01:29:43.000000000","submitted":"2019-05-30 01:29:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658992-1559179783602-45b1259d","meta_rev_id":"5334050641e5ee9e8723d7c39ae2085f9b9a3915","_number":658992,"virtual_id_number":658992,"owner":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-30 01:29:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":2,"date":"2019-05-22 15:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},{"value":2,"date":"2019-05-22 14:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-05-29 23:45:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 09:10:10.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-05-14 12:46:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-22 14:48:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-29 23:45:28.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-05-30 01:29:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39f9925daac32519ba270d5bb8eab2c395fd1d07","author":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"date":"2019-05-14 09:09:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6b0227e6942497e5966c35f041b69c1b2f3a897","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-05-14 09:10:10.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"946695ec1e27e8cd983ba75ea5423430a5760c3a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-14 10:13:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb9380f6f5a536f6a1e20d1ceb913cbc48154a24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 10:54:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/92/658992/1/check/tempest-full/52929a5/ : SUCCESS in 1h 43m 14s\n- tempest-full-py3 http://logs.openstack.org/92/658992/1/check/tempest-full-py3/b0c3ba7/ : SUCCESS in 1h 32m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/658992/1/check/openstack-tox-lower-constraints/5b2b5f1/ : SUCCESS in 6m 21s\n- openstack-tox-pep8 http://logs.openstack.org/92/658992/1/check/openstack-tox-pep8/15e76d3/ : FAILURE in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/92/658992/1/check/openstack-tox-py27/7ea60e5/ : SUCCESS in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/92/658992/1/check/openstack-tox-py35/0069688/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/92/658992/1/check/openstack-tox-py36/75a888b/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/92/658992/1/check/openstack-tox-py37/9bcafc0/ : SUCCESS in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/92/658992/1/check/openstack-tox-docs/b29c580/html/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ba5aad80f2fec11afebf24af673dd502337113e7","author":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"date":"2019-05-14 12:11:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c191e959a06c8210ff2ba52d8ff0089f03edde1e","author":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"date":"2019-05-14 12:13:07.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nI\u0027d also be in favor of {}-style format strings. I used %-style because I thought I saw them before in OpenStack and didn\u0027t want to break consistency. I only tested the {}-version of this patch in production, as this was the state of the patch before I decided to submit it here.\n\nIn my update I also adapted a test to test the format string functionality.","accounts_in_message":[],"_revision_number":2},{"id":"b5c2086417d505626457700266ec0b76685487cc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-14 12:46:17.000000000","message":"Patch Set 2:\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e I\u0027d also be in favor of {}-style format strings. I used %-style\n \u003e because I thought I saw them before in OpenStack and didn\u0027t want to\n \u003e break consistency. I only tested the {}-version of this patch in\n \u003e production, as this was the state of the patch before I decided to\n \u003e submit it here.\n \u003e \n \u003e In my update I also adapted a test to test the format string\n \u003e functionality.\n\nYeah, let\u0027s wait and see what others say but this looks good to me at the moment, anyway","accounts_in_message":[],"_revision_number":2},{"id":"77e41312de4f5b896e0a92ca0fdb6eb72e63fde4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 14:08:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/92/658992/2/check/tempest-full/f7b5cf0/ : SUCCESS in 1h 56m 50s\n- tempest-full-py3 http://logs.openstack.org/92/658992/2/check/tempest-full-py3/8165650/ : SUCCESS in 1h 22m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/658992/2/check/openstack-tox-lower-constraints/6449f05/ : SUCCESS in 7m 43s\n- openstack-tox-pep8 http://logs.openstack.org/92/658992/2/check/openstack-tox-pep8/ec8835c/ : FAILURE in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/92/658992/2/check/openstack-tox-py27/23f6ab8/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/92/658992/2/check/openstack-tox-py35/32e6d69/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/92/658992/2/check/openstack-tox-py36/4a1070a/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/92/658992/2/check/openstack-tox-py37/4f04782/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/92/658992/2/check/openstack-tox-docs/72f9002/html/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"8d92fd9e904e8c5496bc2dd53b4f21dc4b7d50ba","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-21 18:59:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nThere\u0027s lots of invalid format strings the user could potentially put in the config, and no error handling for exceptions from format(). I\u0027m not sure what the right way to handle it should be - fall back on the raw string as the filename? But it seems to me like we should try to handle it gracefully somehow.\n\nLet\u0027s also update the help for the config option to let people know how to use this, and maybe add a release note.","accounts_in_message":[],"_revision_number":2},{"id":"353b44d1880643d58fe7183c2338c1a4103638c4","author":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"date":"2019-05-22 13:18:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3b80904f5a572ac9b13c452426e9b554d4ef1c0a","author":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"date":"2019-05-22 13:22:16.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nHandling broken format strings is a good idea. I now catch exceptions thrown by the format, display a warning and fall back to the original string without formatting. Also added a test for this behaviour. I decided to catch the individual format() exceptions, but we could also change this to just catching Exception to be on the safe side.\n\nI also added a sentence about {pid} to the help of the backdoor_socket_path config option.\n\nIs adding a note about this onto the release notes also something I should do in this commit or is that handled by the oslo team?","accounts_in_message":[],"_revision_number":2},{"id":"92252f2f0e9736755c56dd70fe35074bc7aaa0bc","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-22 13:26:37.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks!\n\nThe best place for a release note is usually in the same commit that made the change. Instructions for creating one are here: https://docs.openstack.org/reno/latest/user/usage.html","accounts_in_message":[],"_revision_number":3},{"id":"038b3b0578f457b370029d14b26d5dc2e80d5003","author":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"date":"2019-05-22 14:22:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"55250e8f7ba82d1d49b0d9551e18b191b09bb853","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-22 14:48:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8ce62ec925fdc6e9f293473880333dab08b4fe3b","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-22 15:57:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f19c8267eb9c840e327d51cb7cd3de04a7c094da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 18:45:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full finger://ze06.openstack.org/510a0e69682542e3a3631798f6cb43f1 : RETRY_LIMIT in 32m 34s\n- tempest-full-py3 http://logs.openstack.org/92/658992/4/check/tempest-full-py3/d57c7a1/ : SUCCESS in 1h 32m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/658992/4/check/openstack-tox-lower-constraints/09ec50d/ : SUCCESS in 6m 34s\n- openstack-tox-pep8 http://logs.openstack.org/92/658992/4/check/openstack-tox-pep8/eeec6d6/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/92/658992/4/check/openstack-tox-py27/0c061c3/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/92/658992/4/check/openstack-tox-py35/34ec18b/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/92/658992/4/check/openstack-tox-py36/f58c233/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/92/658992/4/check/openstack-tox-py37/e8a7ef3/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/92/658992/4/check/openstack-tox-docs/ca53eb6/html/ : SUCCESS in 7m 44s\n- build-openstack-releasenotes http://logs.openstack.org/92/658992/4/check/build-openstack-releasenotes/4dc7709/html/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":4},{"id":"3e877785d906e3a34bc7f6b35312f0f5df744870","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-29 20:14:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7718d05d612745ed642704d5b8ddf6866a7936e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 22:15:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/92/658992/4/check/tempest-full/6fccc14/ : SUCCESS in 1h 51m 46s\n- tempest-full-py3 http://logs.openstack.org/92/658992/4/check/tempest-full-py3/76f6f95/ : SUCCESS in 1h 33m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/658992/4/check/openstack-tox-lower-constraints/3147a6b/ : SUCCESS in 5m 54s\n- openstack-tox-pep8 http://logs.openstack.org/92/658992/4/check/openstack-tox-pep8/2cbdb37/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/92/658992/4/check/openstack-tox-py27/de9dfda/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/92/658992/4/check/openstack-tox-py36/89829cf/ : SUCCESS in 4m 51s\n- openstack-tox-py37 http://logs.openstack.org/92/658992/4/check/openstack-tox-py37/6489d27/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/92/658992/4/check/openstack-tox-docs/fa5a71f/html/ : SUCCESS in 5m 24s\n- build-openstack-releasenotes http://logs.openstack.org/92/658992/4/check/build-openstack-releasenotes/7722194/html/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":4},{"id":"458b5a77ed6aef8126e402a5a0e3e166baababc4","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-29 23:45:28.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2ad55d5b8dbe10744806126b1391e43906ea5beb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 23:45:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1dccf2e0a5f8c7cf6f475820c4e1b4c63c9d5e01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 01:29:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/92/658992/4/gate/tempest-full/2a27f42/ : SUCCESS in 1h 27m 42s\n- tempest-full-py3 http://logs.openstack.org/92/658992/4/gate/tempest-full-py3/afc7f00/ : SUCCESS in 1h 35m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/658992/4/gate/openstack-tox-lower-constraints/bd9663d/ : SUCCESS in 6m 06s\n- openstack-tox-pep8 http://logs.openstack.org/92/658992/4/gate/openstack-tox-pep8/acbeb8f/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/92/658992/4/gate/openstack-tox-py27/6f516f5/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/92/658992/4/gate/openstack-tox-py36/c2a14ce/ : SUCCESS in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/92/658992/4/gate/openstack-tox-py37/e397a38/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/92/658992/4/gate/openstack-tox-docs/19b31c1/html/ : SUCCESS in 4m 21s\n- build-openstack-releasenotes http://logs.openstack.org/92/658992/4/gate/build-openstack-releasenotes/71dc28e/html/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":4},{"id":"a9d0c88a81ae6a2fbcb45ef33432b785fa57b595","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 01:29:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dd174fbfde8c2fa173825d495a724865e3bdc87b","revisions":{"76da379a96ecfac2a34430e585cb2b44d67095d2":{"kind":"REWORK","_number":1,"created":"2019-05-14 09:09:59.000000000","uploader":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"ref":"refs/changes/92/658992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/92/658992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/1"}}},"commit":{"parents":[{"commit":"539885155a7ab24f1cdd9cf9b74e26b3a9d47fc6","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/539885155a7ab24f1cdd9cf9b74e26b3a9d47fc6"}]}],"author":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2019-05-14 08:55:18.000000000","tz":120},"committer":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2019-05-14 09:09:10.000000000","tz":120},"subject":"Make PID availabe as formatstring in backdoor path","message":"Make PID availabe as formatstring in backdoor path\n\nWhen multiple processes are spawned with the same configuration\neach process has the same backdoor_socket path configured and\nonly the first process able to bind to the socket can later be\naccessed via the backdoor. To give each process a unique socket path\nwe now expose the PID of the process as a format string argument,\nwhich can then be used like this:\n\nbackdoor_socket \u003d /var/lib/neutron/backdoor-%(pid)s\n\nChange-Id: I3f86f4867eb0cd5010abadf68620aa3450d3e64d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/76da379a96ecfac2a34430e585cb2b44d67095d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/76da379a96ecfac2a34430e585cb2b44d67095d2"}]},"branch":"refs/heads/master"},"2250fc15ebec68a94fad810d7ce6c7da95a46d6f":{"kind":"REWORK","_number":2,"created":"2019-05-14 12:11:02.000000000","uploader":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"ref":"refs/changes/92/658992/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/92/658992/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/2"}}},"commit":{"parents":[{"commit":"539885155a7ab24f1cdd9cf9b74e26b3a9d47fc6","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/539885155a7ab24f1cdd9cf9b74e26b3a9d47fc6"}]}],"author":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2019-05-14 08:55:18.000000000","tz":120},"committer":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2019-05-14 12:09:02.000000000","tz":120},"subject":"Make PID availabe as formatstring in backdoor path","message":"Make PID availabe as formatstring in backdoor path\n\nWhen multiple processes are spawned with the same configuration\neach process has the same backdoor_socket path configured and\nonly the first process able to bind to the socket can later be\naccessed via the backdoor. To give each process a unique socket path\nwe now expose the PID of the process as a format string argument,\nwhich can then be used like this:\n\nbackdoor_socket \u003d /var/lib/neutron/backdoor-{pid}\n\nChange-Id: I3f86f4867eb0cd5010abadf68620aa3450d3e64d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/2250fc15ebec68a94fad810d7ce6c7da95a46d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/2250fc15ebec68a94fad810d7ce6c7da95a46d6f"}]},"branch":"refs/heads/master"},"199b23fa162d9cb14212f4414fbced4dc6133b9e":{"kind":"REWORK","_number":3,"created":"2019-05-22 13:18:37.000000000","uploader":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"ref":"refs/changes/92/658992/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/92/658992/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/3"}}},"commit":{"parents":[{"commit":"539885155a7ab24f1cdd9cf9b74e26b3a9d47fc6","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/539885155a7ab24f1cdd9cf9b74e26b3a9d47fc6"}]}],"author":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2019-05-14 08:55:18.000000000","tz":120},"committer":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2019-05-22 13:18:12.000000000","tz":120},"subject":"Make PID availabe as formatstring in backdoor path","message":"Make PID availabe as formatstring in backdoor path\n\nWhen multiple processes are spawned with the same configuration\neach process has the same backdoor_socket path configured and\nonly the first process able to bind to the socket can later be\naccessed via the backdoor. To give each process a unique socket path\nwe now expose the PID of the process as a format string argument,\nwhich can then be used like this:\n\nbackdoor_socket \u003d /var/lib/neutron/backdoor-{pid}\n\nChange-Id: I3f86f4867eb0cd5010abadf68620aa3450d3e64d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/199b23fa162d9cb14212f4414fbced4dc6133b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/199b23fa162d9cb14212f4414fbced4dc6133b9e"}]},"branch":"refs/heads/master"},"dd174fbfde8c2fa173825d495a724865e3bdc87b":{"kind":"REWORK","_number":4,"created":"2019-05-22 14:22:29.000000000","uploader":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"ref":"refs/changes/92/658992/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/92/658992/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/92/658992/4"}}},"commit":{"parents":[{"commit":"539885155a7ab24f1cdd9cf9b74e26b3a9d47fc6","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/539885155a7ab24f1cdd9cf9b74e26b3a9d47fc6"}]}],"author":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2019-05-14 08:55:18.000000000","tz":120},"committer":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2019-05-22 14:21:44.000000000","tz":120},"subject":"Make PID availabe as formatstring in backdoor path","message":"Make PID availabe as formatstring in backdoor path\n\nWhen multiple processes are spawned with the same configuration\neach process has the same backdoor_socket path configured and\nonly the first process able to bind to the socket can later be\naccessed via the backdoor. To give each process a unique socket path\nwe now expose the PID of the process as a format string argument,\nwhich can then be used like this:\n\nbackdoor_socket \u003d /var/lib/neutron/backdoor-{pid}\n\nChange-Id: I3f86f4867eb0cd5010abadf68620aa3450d3e64d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/dd174fbfde8c2fa173825d495a724865e3bdc87b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/dd174fbfde8c2fa173825d495a724865e3bdc87b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
