)]}'
{"id":"openstack%2Foslo.service~710798","triplet_id":"openstack%2Foslo.service~master~Idb9b8ca280e5fcb90ab9fe29478dda72bf4a82e0","project":"openstack/oslo.service","branch":"master","hashtags":[],"change_id":"Idb9b8ca280e5fcb90ab9fe29478dda72bf4a82e0","subject":"Update eventlet.","status":"MERGED","created":"2020-03-02 15:37:26.000000000","updated":"2020-03-09 18:51:56.000000000","submitted":"2020-03-09 18:50:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710798-1583779836780-a42b5d9c","meta_rev_id":"503842286261d937eee455e7aaeb261d366f955b","_number":710798,"virtual_id_number":710798,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 18:50:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-03-09 17:29:27.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-03-09 17:19:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-03-09 17:29:27.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 17:19:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-09 17:29:27.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-03-09 18:50:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"454e9366be79003de770c0c2bcceb84b01c4f213","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-03-02 15:37:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3e0df75798ca437f702e5311ea3384d8b1953ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 17:30:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e237d31fbaa841938e89f9e55964f6c7 : SUCCESS in 4m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91b95e095062478c8cd7f2cd2f16f565 : SUCCESS in 1h 45m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40e7867b8f68447995c44e895c4cbb21 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a034bee007bf4fc69bc57417842e2c93 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/308a0af180e746f1afec30fca10d6cbf : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c6fd6e8151247eb9b21cd5a0550c873 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e080f3ca67bb47af96ed6cb729ac0fa5 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/605cc03196e343baba10b297607b6a21 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"872d1c3673aad79114478330d706045c4a0cf830","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-03 11:02:17.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHello,\n\nThanks for these changes.\n\nBefore moving forward with these changes I prefer lets time to other oslo core to discute about them, they have more background history than me about this hot topic.\n\nEventlet versions had break things multiple times in the past so we need to be cautious with this kind of change.\n\nAnyway, by waiting see my inline comments.","accounts_in_message":[],"_revision_number":1},{"id":"9c5b2d61e0b1004a41fb467fa5c47011a8ea359b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-03-06 11:08:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b1aef8c388eee82c1439de2dd6e4c41d9615d5fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 12:46:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/16ab259cb2074a659dded62cf7ad9f28 : SUCCESS in 3m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55482613034b499987e6c39d0ccdd7dc : SUCCESS in 1h 34m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db11c624cdaa4efbb42a8ebf2d92fe7d : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4cacef77cbf4d53a860f5038efef5ce : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9738bef190140a085c2d0348e3c45af : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88c3aac4532747d09ff974bf335fddde : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49992aff59e74958851b740ea3eb39e3 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f1f18daba814164afb6bb6033984d6c : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8fa6e022a83f4509ac96bd4e5aaa00a1894520d8","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-03-06 15:52:52.000000000","message":"Patch Set 2:\n\nIt would be nice to pull in the blacklisting from global-requirements as well while we\u0027re doing this. I realize we were missing some of the newer ones anyway, but since we\u0027re syncing up we might as well get everything.","accounts_in_message":[],"_revision_number":2},{"id":"f779059ea9b3df23b10e205e46285849dc23eca8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-09 13:58:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee4984cc779a70994fadb18faada823332523953","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-03-09 15:12:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1c9ae5e17e857fa135158649266d8f76bc8a9748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 16:26:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/99a4de81018a4011ac4839f1354f03f6 : SUCCESS in 2m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f21ff3aca3a444a8a5d29d2ad61fed8 : SUCCESS in 1h 12m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a776c0ac7884ec388eb72568a352c96 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10ae7b0312f642879be6228e407f8e73 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/943898cab9574950adb524fcfabdf755 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5faef45599af43d9b847e12eece6e27c : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6c5bab5edde461b879fdc6019bce157 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd104ba5bcd4efba8d157e9ba3333a7 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"7487bdb80af49e0fea729689861efb47cb21e6e7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-09 17:19:22.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM,\n\n \u003e It would be nice to pull in the blacklisting from global-requirements\n \u003e as well while we\u0027re doing this.\n\n@Ben: How do you want to pull from the global-requirements? Not sure to see what you want here.\n\n \u003e I realize we were missing some of\n \u003e the newer ones anyway, but since we\u0027re syncing up we might as well\n \u003e get everything.\n\nI think we are ok with these one.","accounts_in_message":[],"_revision_number":3},{"id":"7fa387c048ba58965118de6123caa890dfad6bd0","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-03-09 17:29:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n\u003e LGTM,\n \u003e \n \u003e \u003e It would be nice to pull in the blacklisting from\n \u003e global-requirements\n \u003e \u003e as well while we\u0027re doing this.\n \u003e \n \u003e @Ben: How do you want to pull from the global-requirements? Not\n \u003e sure to see what you want here.\n\nI just meant to make our requirements match the ones in global-requirements. The latest PS does that so I\u0027m happy now. :-)","accounts_in_message":[],"_revision_number":3},{"id":"9619d016581c181d12670d7c07d5833b04fa0820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 17:29:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"475cb990753e020e785e3db4ef62cf49d6608181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 18:50:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aea53e001d0f4546abb0fdd5feb2143a : SUCCESS in 2m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe0c6fb0888240509a9068fa65bc492b : SUCCESS in 1h 20m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a851f70fe2d42a5ae3a8419c2c2cf29 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5abf812c7ede4e1d9e7303bbe58908ad : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49d8d0de4e9141c5bf81d78a515cf0c4 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6b69b3b02ca40f0a8b07643c650db77 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246d3cd821264a4992d7ce25c0cd8684 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"033afb31d5b4aa7a76c8bf83147132c790d72b8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 18:50:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"503842286261d937eee455e7aaeb261d366f955b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 18:51:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb04002872a4794b4f5aa15ef95622c : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"19517d0bd6b34a54ae2779aff98af5172c09e4b6","revisions":{"8a96c805c0cd3f2a794899d9aa8e2b0e0dce9815":{"kind":"REWORK","_number":1,"created":"2020-03-02 15:37:26.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/98/710798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/98/710798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/1"}}},"commit":{"parents":[{"commit":"d2a825b9f8002d320d5038ac3621820b46f48c5c","subject":"remove outdated header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/d2a825b9f8002d320d5038ac3621820b46f48c5c"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-03-02 15:37:09.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-03-02 15:37:24.000000000","tz":60},"subject":"Update eventlet.","message":"Update eventlet.\n\nUpdate eventlet to align with others OpenStack projects.\n\nChange-Id: Idb9b8ca280e5fcb90ab9fe29478dda72bf4a82e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/8a96c805c0cd3f2a794899d9aa8e2b0e0dce9815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/8a96c805c0cd3f2a794899d9aa8e2b0e0dce9815"}]},"branch":"refs/heads/master"},"a62e0f8fb02a2cec7ed61bdc388336efeb147464":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-06 11:08:34.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/98/710798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/98/710798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/2"}}},"commit":{"parents":[{"commit":"ad1d88f895506f0b1d184ae0e499641c2912e517","subject":"Update the minversion parameter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/ad1d88f895506f0b1d184ae0e499641c2912e517"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-03-02 15:37:09.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-03-06 11:08:26.000000000","tz":60},"subject":"Update eventlet.","message":"Update eventlet.\n\nUpdate eventlet to align with others OpenStack projects.\n\nChange-Id: Idb9b8ca280e5fcb90ab9fe29478dda72bf4a82e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/a62e0f8fb02a2cec7ed61bdc388336efeb147464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/a62e0f8fb02a2cec7ed61bdc388336efeb147464"}]},"branch":"refs/heads/master"},"19517d0bd6b34a54ae2779aff98af5172c09e4b6":{"kind":"REWORK","_number":3,"created":"2020-03-09 15:12:24.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/98/710798/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/98/710798/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/98/710798/3"}}},"commit":{"parents":[{"commit":"ad1d88f895506f0b1d184ae0e499641c2912e517","subject":"Update the minversion parameter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/ad1d88f895506f0b1d184ae0e499641c2912e517"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-03-02 15:37:09.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-03-09 15:12:12.000000000","tz":60},"subject":"Update eventlet.","message":"Update eventlet.\n\nUpdate eventlet to align with others OpenStack projects.\n\nChange-Id: Idb9b8ca280e5fcb90ab9fe29478dda72bf4a82e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/19517d0bd6b34a54ae2779aff98af5172c09e4b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/19517d0bd6b34a54ae2779aff98af5172c09e4b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
