)]}'
{"id":"openstack%2Foslo.service~728620","triplet_id":"openstack%2Foslo.service~master~I4f21de08de3bfd89bf8e331b6c819932cde8ca2f","project":"openstack/oslo.service","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"I4f21de08de3bfd89bf8e331b6c819932cde8ca2f","subject":"Switch to newer openstackdocstheme and reno versions","status":"MERGED","created":"2020-05-16 07:52:13.000000000","updated":"2020-05-19 13:36:14.000000000","submitted":"2020-05-19 13:34:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728620-1589895252550-ed83af0c","meta_rev_id":"086d7c64f611a0919a7d71960ef4ca123c97dc9b","_number":728620,"virtual_id_number":728620,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-19 13:34:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-05-18 19:27:59.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-05-19 12:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-05-19 12:19:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 19:27:59.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-05-19 12:19:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-19 13:34:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efabae536d866d845dbb3a63efb38ef729c17939","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-16 07:52:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb60a5ce76dcabe450c7ecb4e90c5accf830570d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 09:11:20.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3c9652f90c64ac2b9daf7ca48173e0c : SUCCESS in 2m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35b9154992f44b649c9bd82765a26fae : SUCCESS in 1h 18m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5a33a45e6f04a489aac8585a2b86706 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/297075133c4543c8ae60fab2da9f4dfa : FAILURE in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f14527a3f843467c9fdfe4c464625ad4 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c781b1dcbe240859de28a708289647e : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43511babb74442aea226ec2ba52b0771 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31f96022a09e4b9a8863aac40ce2a259 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0c1b0a97bd64ac621f0b702eacce8aba66079c75","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-16 17:26:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dafc61e8f98008235ec19e169c8d42c7ee93a4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 18:41:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7bd9c3746cc848d38327fb1c56072fab : SUCCESS in 2m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2acebf7e40b14b3c82c7f0689b30e095 : SUCCESS in 1h 14m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed6654759f7d446b94386d2e4dfeb669 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd7f4152ff3c4da18e6794d4a4fa01a3 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0e70717e20b446aaebb736dae2755e7 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b58a1eef78cf446f802453972084e3f3 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c12247145c64b99bf23e484f8c77e42 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46a040d485b14768991357128be80712 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ee815752f9493f171bd0877f82aa15c6cd50d135","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-18 16:32:52.000000000","message":"Topic changed from reno-openstacksdocstheme to reno-openstackdocstheme","accounts_in_message":[],"_revision_number":2},{"id":"376243bf698eca1411ac9c37a9e2fa42cdde365a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-18 18:53:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8627fbb0d7aedbd9efd2fd7ae16293f69250f2af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-18 18:54:28.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f935d8a15b57a3629354675cbca41e75ae23d21d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-05-18 19:27:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"68787fbce4df0ecb530717c376ca6c7672ec22c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 20:25:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa9e7f6e037e43bab505b82731aa10ad : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc0c7944380c44d38f1c57a9fc1a8eb0 : SUCCESS in 1h 29m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a62641bf9f2d4d79b1bf8c5533b25891 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84f102e849547218eff5dbb3c2a682a : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e7620095b944b29bae4651bfcff032d : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82ba94ebd6f942ba9f5b7557bc1bacf9 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/227d986b22f14e6a99a509566323ffba : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1f33f60c7d547fd8e795b3d347d42b4 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":4},{"id":"78e78eb2721e99542046403490660afd0975fa3b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-19 12:19:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"81ba7ce0ace1f28746fe0d39652242c6b0c481ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 12:19:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9686ba293f14a6dfe9df28c1ffbb53e67b3c1b64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 13:34:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1e20da36f73d4b7585e51f0b21b51c16 : SUCCESS in 3m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1aced8d3912741c8bccbdeb3a0b0c2f1 : SUCCESS in 1h 12m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/796e23c1dfd3499b8b66a0e7fa83abd0 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa5c92a16714fb2baba51ef0a0ecfe7 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83ad03c3657849d590528bbd077d5728 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15c763ff103847d7a3007810f02fd94f : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2bae1d3c8034f4b86dca3dad47e687a : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e899c1c583c4de7a6595e24e96f64df : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"a4b5c23fed4c8f2e8cd21b82b1373685b2429ea4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 13:34:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"086d7c64f611a0919a7d71960ef4ca123c97dc9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 13:36:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/329549b1a2fc424e8812e080ef13e7bb : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f4a4b6330af447cae975e8414038b6c : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3803ff29db71f3c4b3dd719979efa646573fbfb3","revisions":{"9e78fccd9d232bd06ca2a1e9ce474fc2ad111723":{"kind":"REWORK","_number":1,"created":"2020-05-16 07:52:13.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/20/728620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/20/728620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/1"}}},"commit":{"parents":[{"commit":"38306c8319eea4301aa37472bd9dc8a32780ca1c","subject":"Merge \"Remove monotonic usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/38306c8319eea4301aa37472bd9dc8a32780ca1c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-16 07:51:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-16 07:51:57.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.1.2 and reno 3.1.0 versions. Using\nthese versions will allow parallelizing building of documents.\n\nUpdate Sphinx version as well.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nChange-Id: I4f21de08de3bfd89bf8e331b6c819932cde8ca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/9e78fccd9d232bd06ca2a1e9ce474fc2ad111723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/9e78fccd9d232bd06ca2a1e9ce474fc2ad111723"}]},"branch":"refs/heads/master"},"9ec12afecf5623d7787132faa905685c3c50501b":{"kind":"REWORK","_number":2,"created":"2020-05-16 17:26:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/20/728620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/20/728620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/2"}}},"commit":{"parents":[{"commit":"38306c8319eea4301aa37472bd9dc8a32780ca1c","subject":"Merge \"Remove monotonic usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/38306c8319eea4301aa37472bd9dc8a32780ca1c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-16 07:51:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-16 17:26:22.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.1.2 and reno 3.1.0 versions. Using\nthese versions will allow parallelizing building of documents.\n\nUpdate Sphinx version as well.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nChange-Id: I4f21de08de3bfd89bf8e331b6c819932cde8ca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/9ec12afecf5623d7787132faa905685c3c50501b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/9ec12afecf5623d7787132faa905685c3c50501b"}]},"branch":"refs/heads/master"},"6e4fa6ab8d8404845ab557b47cb0ce7eaad63b59":{"kind":"REWORK","_number":3,"created":"2020-05-18 18:53:23.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/20/728620/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/20/728620/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/3"}}},"commit":{"parents":[{"commit":"38306c8319eea4301aa37472bd9dc8a32780ca1c","subject":"Merge \"Remove monotonic usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/38306c8319eea4301aa37472bd9dc8a32780ca1c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-16 07:51:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-18 18:52:21.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering\n\nUpdate Sphinx version as well.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nChange-Id: I4f21de08de3bfd89bf8e331b6c819932cde8ca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/6e4fa6ab8d8404845ab557b47cb0ce7eaad63b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/6e4fa6ab8d8404845ab557b47cb0ce7eaad63b59"}]},"branch":"refs/heads/master"},"3803ff29db71f3c4b3dd719979efa646573fbfb3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-18 18:54:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/20/728620/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/20/728620/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/20/728620/4"}}},"commit":{"parents":[{"commit":"38306c8319eea4301aa37472bd9dc8a32780ca1c","subject":"Merge \"Remove monotonic usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/38306c8319eea4301aa37472bd9dc8a32780ca1c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-16 07:51:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-18 18:54:17.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering\n\nUpdate Sphinx version as well.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nDepends-On: https://review.opendev.org/728938\nChange-Id: I4f21de08de3bfd89bf8e331b6c819932cde8ca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/3803ff29db71f3c4b3dd719979efa646573fbfb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/3803ff29db71f3c4b3dd719979efa646573fbfb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
