)]}'
{"id":"openstack%2Foslo.service~730168","triplet_id":"openstack%2Foslo.service~master~I66d8f944a16ba72f4ce934eaf62ae5f03ab2aac4","project":"openstack/oslo.service","branch":"master","hashtags":[],"change_id":"I66d8f944a16ba72f4ce934eaf62ae5f03ab2aac4","subject":"Fix hacking min version to 3.0.1","status":"MERGED","created":"2020-05-22 02:53:59.000000000","updated":"2020-05-26 15:49:08.000000000","submitted":"2020-05-26 15:45:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730168-1590507949049-3f6755ef","meta_rev_id":"5548f85c9a71dabd20922c969d88c4c861852a87","_number":730168,"virtual_id_number":730168,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 15:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-26 14:09:38.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-25 14:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-26 14:09:38.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 14:55:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-26 14:09:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-26 15:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f88a11c411a46bd48e0c1cf9500893410c251989","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-05-22 02:53:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3f140ac3270b4a22a4c624ca0e125725eb95a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 04:22:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aaaba3df5a544ba394e98425d5f1c118 : SUCCESS in 2m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a746ef8c47546f08885afb11f168c01 : SUCCESS in 1h 26m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aba022f08adc4085a88bc33be7567426 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a352f99224554fe28eb67d6abb0dee5d : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3a6b780d7734622a4e4a0daa1b7809b : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3e32e167e564d4ba7b0cd7ddd92fbdf : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a26a14ad504a3ca4ac702d6092a775 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6df8e39c0d148e1a5e2c7200b616a17 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c2284467a2e3c6a7aef819984bcdc0927e3a2ed7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-25 14:55:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b32736ef62ab59d524233aa0e046a6c9e83226b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-26 14:09:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fb275f0f9a89ac2729dddcbd2d63cf4cd18c3df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 14:09:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44144fe266e233c35e361e09d9349e86103d14d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 15:45:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/79542c1443774e869dbd81508b5ab4e1 : SUCCESS in 3m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab3ed95ef96844bb8cdcf1f522bbf276 : SUCCESS in 1h 29m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4791f596de5b451aae7d7ebbc430edcf : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cba3ed0a88e41eea54ed72985bfaaad : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fa87d82aef24a2d9f39d88967c4200c : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad92ab24f08a4e948d80579cf92c616e : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345b73d593c74d6bbabed31fa08b2388 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/422befd3f94c4a99975ee3c2b94be844 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5d7a3af24e9895dbcea060f1d21b2ac7517fadd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 15:45:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5548f85c9a71dabd20922c969d88c4c861852a87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 15:49:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2d768eb2d44edaa7e3cecfbae1df69 : SUCCESS in 2m 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1cba21aae624bdca720c44a8a309249 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e0ef272f8d154224d69aca44563cbc4e2b66160","revisions":{"2e0ef272f8d154224d69aca44563cbc4e2b66160":{"kind":"REWORK","_number":1,"created":"2020-05-22 02:53:59.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/68/730168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/68/730168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/68/730168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/68/730168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/68/730168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/68/730168/1"}}},"commit":{"parents":[{"commit":"3803ff29db71f3c4b3dd719979efa646573fbfb3","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/3803ff29db71f3c4b3dd719979efa646573fbfb3"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-05-22 02:53:33.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-05-22 02:53:52.000000000","tz":480},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nChange-Id: I66d8f944a16ba72f4ce934eaf62ae5f03ab2aac4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/2e0ef272f8d154224d69aca44563cbc4e2b66160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/2e0ef272f8d154224d69aca44563cbc4e2b66160"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
