)]}'
{"id":"openstack%2Foslo.service~836637","triplet_id":"openstack%2Foslo.service~stable%2Fstein~Ia531606ec2526ce1c6169296a8f254342870416d","project":"openstack/oslo.service","branch":"stable/stein","topic":"oslo_lc_drop","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-29 09:58:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia531606ec2526ce1c6169296a8f254342870416d","subject":"Dropping lower constraints testing","status":"ABANDONED","created":"2022-04-05 11:31:07.000000000","updated":"2022-07-29 09:58:16.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"60d286ed580d048e9163e519795351555b8f2702","_number":836637,"virtual_id_number":836637,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-05 13:04:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-04-05 11:33:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-04-14 14:06:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-04-05 11:33:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 11:31:07.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-04-05 13:04:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-14 14:06:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"af903f000eec281b2603d29b05a3d961d7306708","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-05 11:31:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8df55e8f7b63ba1605050df42475b407c8bc1c91","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-05 11:33:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7215ef09e8a840b329dc78ff020ca8aa63c64f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:04:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/67c7b07e8b414508b09363c956e5bc8b : SUCCESS in 3m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/194153b3d10e48db87c8f0ffd558dd64 : SUCCESS in 1h 19m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62a113ca2aa843f7957fc42f94411dac : SUCCESS in 57m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b18ca880a4a04a1b83d4984fe6433282 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e775cd2306643deb3b001bf3c9e902d : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5197a9729b704f26baea54657802b3cd : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1289cd94f0d84ef9a161d59995243b36 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebb3400768a048a6bfcbc6a8ec2a3bda : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90f07857c0a84a22a4d94179b7631cfb : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb8992c562f24b63a8222655cc12fe31 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"cb350ade60bdf68246ab9123dc0817b905144794","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-05 13:12:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23ef5b50395effc56bc5888ec2ea63f57712a431","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-05 13:12:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efd30eaf7551a02e13cad3bfc559eb89f7ad9c8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 14:36:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b781fa97c3534319a9f39528076eb337 : SUCCESS in 2m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/83c614b88d294eb9a53c474762aa43ec : SUCCESS in 1h 17m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89342c297d8240db826ae0ac491ed700 : SUCCESS in 1h 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13c676b2e90b4c80b1911abd80964a6d : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92e7de71227741c29f3c511f2f5dd387 : SUCCESS in 4m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c2613c0427584bf9aeae7bc5077591b5 : FAILURE in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f7a380afb7b421d910354999999b9df : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a64397ec41547bfa00992d6587470e4 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af2a8b4d6e9453383929b90a68a93a5 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a4276984404468d809691f5f3332b54 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1992217ce6113dcfe8a001534dfe5f8ba51fe0a6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-05 15:23:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c53fe9fa69124ebf27ced6b8855cac75fb65ae39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 17:18:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/74847ff09fcd4503bcfa0059cf09be48 : SUCCESS in 3m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5635e8f76255480da5778f4c1ca41acc : SUCCESS in 1h 26m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bdfb473999a4181a2ba5069c7d272fa : FAILURE in 1h 41m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30172bab07954c79bfe8b3584fca8956 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ec92a7f30a74a98ae67979e46b3c911 : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4ea2ca9c30a64be8baa7b4eb83b2123f : FAILURE in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/016691364da54e5e9d9cc9173378dc1d : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9da35d88c05944e8a7b4eaeb4f14ba2f : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e424a6828b74b12a0d3ee926da9c3a6 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee3a59d29a0f4852ab00a020d91b6f8c : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b97004942111b1cf89bcf48ae75f9d408fa4b217","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-07 10:49:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07610f694c8416deb0ac688033d0a634c5289c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 12:09:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ab6f63d6e8ab480b8fca622f270172ca : SUCCESS in 2m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2315e61d245841cb87a2cc1206aa0f68 : SUCCESS in 1h 13m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fae82527bcbb407abad5dd8e8b153793 : SUCCESS in 1h 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb12624a924f4c7da412b991a68f885d : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f8c6a92540d4626bed7e6c10d74d3f9 : SUCCESS in 3m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/39c151ea1eb0468296102096ff596c9c : FAILURE in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b97e65a0f904fc686cc9623c6616eff : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2719509721fb4b3291843ba0fce7e241 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ca322630ed493c8e7bbf0e10c2fcd6 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c64903af1844dcbb7d0f448efd5d088 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"653eb0d5b30307eda3275a4c281137c32863769b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-14 14:06:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac631de82780ce5edd77f466949d282f4e12d70d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 14:18:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e350a4805494e89a839d9c3e2761264 : SUCCESS in 2m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b1ea569b1c2f4b02b4feb4d2bfa90bd5 : FAILURE in 9m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/617e106ae4ab480b86ec1789c178a3c2 : FAILURE in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffba85cd1dba442a96016c2e596cde54 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17adc64a8ab14ccab4108a4106071d25 : SUCCESS in 3m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8707c24495314e25ad6e2a84d90398df : FAILURE in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/977c25ac487b411589dc339c9d3ce9fe : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82fa012a1dec4aa4836cf139e1100384 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8333d4f6f419477bacff8882f0185ca7 : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/425c1fc2dc2a434fb72e46c8f1abe2d6 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"60d286ed580d048e9163e519795351555b8f2702","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-29 09:58:16.000000000","message":"Abandoned\n\nThis branch is clearly broken. I think it\u0027s time to just close off stein","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06fb18e4b09bc8f0007e8bccb1d4e992a3f232cd","revisions":{"06fb18e4b09bc8f0007e8bccb1d4e992a3f232cd":{"kind":"REWORK","_number":1,"created":"2022-04-05 11:31:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/836637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/37/836637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/37/836637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/37/836637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/37/836637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/37/836637/1"}}},"commit":{"parents":[{"commit":"823139677728d89da7216f9e5291b0371a6ce0cd","subject":"[stable-only] Cap bandit and fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/823139677728d89da7216f9e5291b0371a6ce0cd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-18 14:12:43.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-04-05 11:30:43.000000000","tz":60},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain\nneeded to maintain that now exceeds the benefits provided by this\nmechanisms.\n\nConflicts:\n  lower-constraints.txt\n\nNOTE(stephenfin): Conflicts are due to the different lower constraints.\n\nNOTE(stephenfin): Change Ibe9e1aa7501e80844ef4b4e05866c0c3cf310c86 is\nsquashed into this backport.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: Ia531606ec2526ce1c6169296a8f254342870416d\n(cherry picked from commit a88653d07787d1ea29ce7bd05bf73db42db48c94)\n(cherry picked from commit bdcab45915043c8e031b967553b3d5e96600a6b4)\n(cherry picked from commit 839022d747a66c7899a85c2111d8ea0de4618b0c)\n(cherry picked from commit 1be3ac39984c579d7a8d4f0994d02d311b6e948d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/06fb18e4b09bc8f0007e8bccb1d4e992a3f232cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/06fb18e4b09bc8f0007e8bccb1d4e992a3f232cd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
