)]}'
{"id":"openstack%2Foslo.service~975574","triplet_id":"openstack%2Foslo.service~master~I412c7476f2765d3c95952af441005a97a3e4a0bc","project":"openstack/oslo.service","branch":"master","topic":"bp/nova-services-graceful-shutdown-part1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-06 21:50:43.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2026-02-09 08:36:03.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-02-12 07:07:19.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-02-12 03:53:22.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2026-02-12 07:07:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I412c7476f2765d3c95952af441005a97a3e4a0bc","subject":"Add register and set method to override the default value for service opts","status":"MERGED","created":"2026-02-03 20:47:31.000000000","updated":"2026-02-12 07:08:11.000000000","submitted":"2026-02-12 07:07:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975574-bp/nova-services-graceful-shutdown-part1","meta_rev_id":"b7330f6764647a466dcb649463412668635a05fb","_number":975574,"virtual_id_number":975574,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 07:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-02-06 21:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-02-09 08:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-02-09 08:36:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-03 21:48:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 06:33:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-02-05 11:46:50.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2026-02-05 13:13:56.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2026-02-06 10:16:05.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-02-06 13:46:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2026-02-09 08:36:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"08c7d13c381cff40147b9b5904ccc00164816709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-03 20:47:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47ede44ab5c4a71d7e2c9ab7d452eac05780d904","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-03 20:48:41.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed083b294f460c59a03faa09d9e17331cbd1b92f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 21:48:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dbed294389a4937bb2ce0b067ba404f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5a020f3a4a349b6ac265d69cb4f5aeb : SUCCESS in 1h 00m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de8068babb354ef6a2a79e90edb8c7de : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1e0bef15c1647e5a4dd89d673b0ddbe : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0e0680c643d4454a91512cb735ec94f : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a6dac88a0ef43e8bc901edc427e4c96 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddfbce714997471abfec1610ec54bb5d : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b090cf66fffbfef005409d6134e74d3587fdbf97","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-05 04:52:50.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"170fd11f366f1ed8c78c9f423a11c860e68a0959","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-05 06:33:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"533b2b3de523d1f4383ac14f3249fb7561560551","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-02-05 11:46:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d34fb52efff1d0ca04b1e30c41f6d63b2c6053d0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-02-05 13:13:56.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"60948aee1f2ac3799be3e02fb6ac777d71b014ab","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-05 13:59:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fdc162a3e5eb9147f63b0bfbeb682faf5a59595","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-05 14:53:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45819dd3ab7a38b81b977455797b8b5245e8663f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-05 19:12:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78a31807d872e9ab36c4686dbfa20a6c3b82dcf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-05 19:20:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"794ccc4645e9792e8cd0d72b478f1ba9b1cf1fa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-05 19:23:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"99fbb330f050afcd31cd61bb9d0b055b097eda2a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-05 19:23:37.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1d79cb7de4eb5c21fa6d620b2340e6b30f3a68c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 20:26:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9ff14d42977481787456d6c8a64580b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a851e76935974e1dbabd209da056a3a8 : SUCCESS in 1h 02m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e0669260203480ea44779f27a021f9f : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c91012de6af141628ffa6e08f030c771 : FAILURE in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b284ada6ff424621ad8ce269da4a5a40 : SUCCESS in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072b9180600c4cd5b5737e5511f70ed6 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d76b6b293f94ca09930c7f8034954f4 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":3},{"id":"40721af7559d0c77da96e691b88ff0bb4c21f0d8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-05 21:12:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cda4498ffc40176b5e3bd962d8e7180e83489a51","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-05 21:17:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"549bd87cdb29a80e705192c244ac9a50af77ab39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 23:08:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd433b6599dc4288842926d1360b8acb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/374d887f9fc44854aef5a5e983161e7b : SUCCESS in 1h 51m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f09f226d2e564faf9c15a07309d1e452 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2286d624d2d9473bade828949b0a5251 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b2215a712b44a31921589154070d388 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514733c27ae84bb29c2695f21fc05e0c : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83ab97228dfa4c52a55ed1f4ec1d1dd3 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":3},{"id":"eb392efa5a99b403b09f8cb852c04e321c7787ae","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-02-06 10:16:05.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"38402bae881d7bbdc4df21e4feecbc0c953eb7dc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-06 12:11:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70d70a4b18cc54e9b69857b98a601e4f5434a43a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-02-06 13:46:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9603add7e6b0a57c26ea345275154c655c0709d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-06 18:12:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a63e35ea0b1082c7bc1e7dbb369e3ea2ff5427ea","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-06 18:12:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8fc745a59599efe2d1c5178c0ecf8967eb139c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 20:15:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b32f2c9e973f4bc98498110c439c1eef\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05f8cd7911ab484e816e27883de9195a : SUCCESS in 1h 57m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d3bac41a73e4b7294521878f2d55078 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad39a51a12344218a67d9deff13e36dc : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/638c896e52374842b5afcfb5cb0c3b75 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d066d479e8454ba45f23895477e543 : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3aaf2a7ca3d343f3a08ea9a3aeb7c61e : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":4},{"id":"35527da84e438b608d0d4649fdef358bbbf5f5d1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-06 21:50:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"74545fb5de7eb0e2e4fa8fd62354117271a873ab","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-02-09 08:36:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7eaed376f8bd94a980b5c6463b356084c2df391d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 08:36:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0a73960e43f061ac1d2129924937afd23459100f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 08:50:21.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bb13e901dae4f9d922f360eb8f39cda\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9225f566dbbf432eb9113c08a3737b5f : FAILURE in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8219882a28384b0097a0131b2578a012 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/636e6f756ce5492fb3d18f2f30e54f8e : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc60673730b74c0dbbaa5c2e7fe979f4 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba1186a66afb466596ee7a21e99c8247 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b0deda8eb674144ba501ad17519c87c : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":4},{"id":"2cf477b361d8064259e80eccce311eefe278278a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-11 15:47:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"922a20615cd5981201530b93b1d303166313ce8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 16:20:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1169fca4c5b54c37a1a67a07e8a547be\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b69ade7736b748569c5c8284891fddde : FAILURE in 27m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5cb831d450f41d1a32a57208a0749dd : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ac7d1b642804a17974d1d414e15668e : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b33dd269de54ad1a210cb141fbbdcff : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a99874296241548939ced0ba766d29 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86cc77873ac54c14906879de235d2273 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":4},{"id":"8d6a4853a92b71b6e87b1ffd4041c988c5bf742a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-12 03:53:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5ab79ce06e214cb795e47a5ec0810de24445d0ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 05:02:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf3adb5276f4b36a08f607f56e641a1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b03ee5395834352a6bcfb3182aef0be : SUCCESS in 1h 05m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a366ef85a64f31b04ba61e7f6ee162 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9721012120244449a269f3c9b6eb8848 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95c4825390e9432f9ab22e9550c889f5 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c00cf3452d0341abb8743d71dec6466c : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f4c7db37d5d4737af6fe11647f6de75 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":4},{"id":"1f74d74f2347a663fdfd28cb012d2f675402f242","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 05:02:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"db1a4c2cccea1bd74562bd0f50de4d28d10dbd9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 07:07:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0f632f327714792a515eaad9ea6c4dc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f266dd9cb05f4f9f905c5c37628671c8 : SUCCESS in 1h 59m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e389616227b34337b93465ef7f98b362 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0bce1ab91e449c2b330e810d48d000b : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41999f54ed6840a58fdb867f9aea4395 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b8deb67e5ab431297c3916aed5fdc45 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd6593e6fbcd4ef89b87a60f1477bd0d : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":4},{"id":"c678e3ee54bbcbe372e1765a86172f8afebd3dc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 07:07:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b7330f6764647a466dcb649463412668635a05fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 07:08:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdc266abf0fe49b79a851251cb7cd796\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea3024b52674dce94eb9f15390c4986 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee49564209cf4ba284ce5a2c3b7f338c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7c457ffdd80e909f4f3c0875fdcb983c92d6d624","revisions":{"7f3bba9262566f70830c6f940ae8dcc562abb18f":{"kind":"REWORK","_number":1,"created":"2026-02-03 20:47:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/975574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/74/975574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/1"}}},"commit":{"parents":[{"commit":"828aed5421572f0a669e683ff254432e9a276e29","subject":"Add signal handler for ProcessLauncher with no_fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/828aed5421572f0a669e683ff254432e9a276e29"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-03 20:44:50.000000000","tz":-480},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-03 20:44:50.000000000","tz":-480},"subject":"Add a set method to override the default value for service opts","message":"Add a set method to override the default value for service opts\n\nThe service opts \u0027graceful_shutdown_timeout\u0027 is per services option and\nsometime services may need the different default timeout. For example,\nnova needs the graceful_shutdown_timeout default to 180 sec[1]. This\nwill allow services to override the default as per their need.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/2026.1/approved/nova-services-graceful-shutdown-part1.html#graceful-shutdown-timeouts\n\nChange-Id: I412c7476f2765d3c95952af441005a97a3e4a0bc\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/7f3bba9262566f70830c6f940ae8dcc562abb18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/7f3bba9262566f70830c6f940ae8dcc562abb18f"}]},"branch":"refs/heads/master"},"de530b27a73d3d25da30017118bd0eb105bef2ba":{"kind":"REWORK","_number":2,"created":"2026-02-05 19:20:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/975574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/74/975574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/2"}}},"commit":{"parents":[{"commit":"828aed5421572f0a669e683ff254432e9a276e29","subject":"Add signal handler for ProcessLauncher with no_fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/828aed5421572f0a669e683ff254432e9a276e29"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-03 20:44:50.000000000","tz":-480},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-05 19:20:32.000000000","tz":0},"subject":"Add register and set method to override the default value for service opts","message":"Add register and set method to override the default value for service opts\n\nThe service opts \u0027graceful_shutdown_timeout\u0027 is per services option and\nsometime services may need the different default timeout. For example,\nnova needs the graceful_shutdown_timeout default to 180 sec[1]. This\nwill allow services to override the default as per their need.\n\nAlso, add a public register_service_opts method to register the service opts\nso that service like Nova[2] can remove their hack.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/2026.1/approved/nova-services-graceful-shutdown-part1.html#graceful-shutdown-timeouts\n[2] https://github.com/openstack/nova/blob/75aed9a19dcf8f26fcb9e84db7a0696acefd56d6/nova/api/openstack/wsgi_app.py#L121\n\nChange-Id: I412c7476f2765d3c95952af441005a97a3e4a0bc\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/de530b27a73d3d25da30017118bd0eb105bef2ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/de530b27a73d3d25da30017118bd0eb105bef2ba"}]},"branch":"refs/heads/master"},"a0aaac4ceb17edd2f24684bb53a9d0d41c2085a4":{"kind":"REWORK","_number":3,"created":"2026-02-05 19:23:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/975574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/74/975574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/3"}}},"commit":{"parents":[{"commit":"828aed5421572f0a669e683ff254432e9a276e29","subject":"Add signal handler for ProcessLauncher with no_fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/828aed5421572f0a669e683ff254432e9a276e29"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-03 20:44:50.000000000","tz":-480},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-05 19:22:52.000000000","tz":0},"subject":"Add register and set method to override the default value for service opts","message":"Add register and set method to override the default value for service opts\n\nThe service opts \u0027graceful_shutdown_timeout\u0027 is per services option and\nsometime services may need the different default timeout. For example,\nnova needs the graceful_shutdown_timeout default to 180 sec[1]. This\nwill allow services to override the default as per their need.\n\nAlso, add a public register_service_opts method to register the service opts\nso that service like Nova[2] can remove their hack.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/2026.1/approved/nova-services-graceful-shutdown-part1.html#graceful-shutdown-timeouts\n[2] https://github.com/openstack/nova/blob/75aed9a19dcf8f26fcb9e84db7a0696acefd56d6/nova/api/openstack/wsgi_app.py#L121\n\nChange-Id: I412c7476f2765d3c95952af441005a97a3e4a0bc\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/a0aaac4ceb17edd2f24684bb53a9d0d41c2085a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/a0aaac4ceb17edd2f24684bb53a9d0d41c2085a4"}]},"branch":"refs/heads/master"},"7c457ffdd80e909f4f3c0875fdcb983c92d6d624":{"kind":"REWORK","_number":4,"created":"2026-02-06 18:12:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/975574/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/74/975574/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/74/975574/4"}}},"commit":{"parents":[{"commit":"131c4624360eeff891650c64cb0b89aca2ab0140","subject":"Fix ServiceManager singleton race condition in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/131c4624360eeff891650c64cb0b89aca2ab0140"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-03 20:44:50.000000000","tz":-480},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-06 18:12:35.000000000","tz":0},"subject":"Add register and set method to override the default value for service opts","message":"Add register and set method to override the default value for service opts\n\nThe service opts \u0027graceful_shutdown_timeout\u0027 is per services option and\nsometime services may need the different default timeout. For example,\nnova needs the graceful_shutdown_timeout default to 180 sec[1]. This\nwill allow services to override the default as per their need.\n\nAlso, add a public register_service_opts method to register the service opts\nso that service like Nova[2] can remove their hack.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/2026.1/approved/nova-services-graceful-shutdown-part1.html#graceful-shutdown-timeouts\n[2] https://github.com/openstack/nova/blob/75aed9a19dcf8f26fcb9e84db7a0696acefd56d6/nova/api/openstack/wsgi_app.py#L121\n\nChange-Id: I412c7476f2765d3c95952af441005a97a3e4a0bc\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/7c457ffdd80e909f4f3c0875fdcb983c92d6d624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/7c457ffdd80e909f4f3c0875fdcb983c92d6d624"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
