)]}'
{"id":"openstack%2Foslo.service~980440","triplet_id":"openstack%2Foslo.service~master~I58f3fd206eaf17f81158d58daa7ab88e758e9ee1","project":"openstack/oslo.service","branch":"master","topic":"timeout-stop","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2026-03-31 07:46:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I58f3fd206eaf17f81158d58daa7ab88e758e9ee1","subject":"Add get_timeout_stop_sec()","status":"ABANDONED","created":"2026-03-13 14:33:45.000000000","updated":"2026-03-31 07:46:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"17cc83ebeda70aec6576280baa3b48f11ee63ca6","_number":980440,"virtual_id_number":980440,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-27 09:23:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 16:34:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad92ec78be5e3d66c58ffbc767ba6733a29d83f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-03-13 14:33:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2fbf6bf0e7b7ef71b7053b76caa168d259031f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 16:34:55.000000000","message":"Patch Set 1: Verified-1\n\n(21 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da71b31bcccc4185915efc763b5dd127\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5952069b1d841d3a8cdfd72cb327cc2 : SUCCESS in 1h 55m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dcf1eb03c64e50b59612acb48770b9 : FAILURE in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2654409a7fa4fcb91a644103a03b513 : FAILURE in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe3df1957f074ce08a2121a5fb718d56 : FAILURE in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8702c3d5290244dda01a4f5fe85fff77 : FAILURE in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9815681108634473b8ac29fbd9f5f9b2 : FAILURE in 4m 27s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/10a0dfaea39441108fd290ec3b7b45f0 : SUCCESS in 1h 16m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f0a21007805b5bda34e1668ef1feae02f0a3af3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-03-27 08:05:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"20d2b298446d3e37043b9d471821bdf9eb55f5ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 09:23:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64b0e2e143464d369fc0f73ed811b59d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0033ccbd864a47b89307c4e4748d421f : SUCCESS in 1h 06m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b19abb04d499413fae9fdba71e70dba8 : FAILURE in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95cdd9477d9f4c4daea3dd352ca6fa8b : FAILURE in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94ceef2c153f419e94d8e7269ec23b99 : FAILURE in 3m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6614e8946cf34391b564a8bf73d7e0d1 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711d8904dadf46a6924d8208d0972653 : FAILURE in 3m 48s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/0909b1db5ec846ef99c2a9cdea821f58 : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":2},{"id":"17cc83ebeda70aec6576280baa3b48f11ee63ca6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-03-31 07:46:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b4044d2a32a481a60fcbc5088b34fec268ef476","revisions":{"bd799d9f8362846ae3cfed65461914d04544d422":{"kind":"REWORK","_number":1,"created":"2026-03-13 14:33:45.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/40/980440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/40/980440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/40/980440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/40/980440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/40/980440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/40/980440/1"}}},"commit":{"parents":[{"commit":"ebd4a18310ad3ceed3e8a0a6aa9df9dfc16f39ef","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/ebd4a18310ad3ceed3e8a0a6aa9df9dfc16f39ef"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-13 14:32:08.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-13 14:33:31.000000000","tz":60},"subject":"Add get_timeout_stop_sec()","message":"Add get_timeout_stop_sec()\n\nAdd a function to get the TimeoutStopUSec configured in systemd\nfor the current service. This may be useful for example in Nova\nto know what value can be set for graceful_shutdown_timeout.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I58f3fd206eaf17f81158d58daa7ab88e758e9ee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/bd799d9f8362846ae3cfed65461914d04544d422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/bd799d9f8362846ae3cfed65461914d04544d422"}]},"branch":"refs/heads/master"},"0b4044d2a32a481a60fcbc5088b34fec268ef476":{"kind":"REWORK","_number":2,"created":"2026-03-27 08:05:46.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/40/980440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/40/980440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/40/980440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/40/980440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/40/980440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/40/980440/2"}}},"commit":{"parents":[{"commit":"ebd4a18310ad3ceed3e8a0a6aa9df9dfc16f39ef","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/ebd4a18310ad3ceed3e8a0a6aa9df9dfc16f39ef"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-13 14:32:08.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-27 08:05:39.000000000","tz":60},"subject":"Add get_timeout_stop_sec()","message":"Add get_timeout_stop_sec()\n\nAdd a function to get the TimeoutStopUSec configured in systemd\nfor the current service. This may be useful for example in Nova\nto know what value can be set for graceful_shutdown_timeout.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I58f3fd206eaf17f81158d58daa7ab88e758e9ee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/0b4044d2a32a481a60fcbc5088b34fec268ef476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/0b4044d2a32a481a60fcbc5088b34fec268ef476"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
