)]}'
{"id":"openstack%2Foslo.service~994645","triplet_id":"openstack%2Foslo.service~master~I9fe1dfe6bf139dc4e422d1b5ba4f7e8fe5f01d54","project":"openstack/oslo.service","branch":"master","topic":"python-next-jobs","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-25 18:39:27.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I9fe1dfe6bf139dc4e422d1b5ba4f7e8fe5f01d54","subject":"zuul: Use openstack-python3-next-jobs template","status":"MERGED","created":"2026-06-24 10:18:51.000000000","updated":"2026-06-25 20:25:40.000000000","submitted":"2026-06-25 20:23:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994645-python-next-jobs","meta_rev_id":"01c8c7017e20e7691481ae122fc05b6ceb21df11","_number":994645,"virtual_id_number":994645,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 20:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-25 18:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-25 18:39:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 12:27:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61830868f9ebd000070422a08c846e623bc3c5b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 10:18:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2404f89f306c8620931af6f9f18e1b0ef91d04b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:27:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a26fef58803a449da2d2275a352fdc9a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/165e264a8b5745ffb74213b59111f59c : SUCCESS in 1h 53m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8d6420f8274dacbebcc51b06b9139d : FAILURE in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e90e28fec0d2478aa3844099a47467db : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8b142a76afd4bffaad95fee3ccf0f1d : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5a1444d1b34a4b5b9c805b708df6ef8b : SUCCESS in 6m 42s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/e821084115164079a5e7509cab997b1f : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26945d61adb41f6824217701f1c8d43 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/422af552d38e4ba18121207491f63c1a : SUCCESS in 3m 47s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/a63ac53fd86245d699a08c8be70b7246 : SUCCESS in 1h 09m 57s\n\nWarning:\n  Comments left for invalid file oslo_service/backend/__init__.py","accounts_in_message":[],"_revision_number":1},{"id":"8978ccb9dd0dea101e813323758d7d336e782cab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 12:48:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e093cf0ddf009d7851378de0002769a99d065465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:30:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797032d331b8417db482463182068a20\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aad05e56ffef42ab9795b1837ff87f79 : SUCCESS in 1h 35m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e362c2c3d749ddbbae3af1ad6fce67 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bef62b61434d493f9c7217488b70fe4d : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bb0af427e7e41c4a808945ad2c79d41 : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2474a40c7e00449d912d73a7c5fa2e11 : SUCCESS in 6m 34s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4aa6d726a7e94986876a3c1cea477aa9 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f43d1e038d2494194dd5ebea99f7af8 : SUCCESS in 2m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3076b0c9b634ce38b7d0462749e3da3 : SUCCESS in 2m 50s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/30233337afd343b4a7e6ecce28a74ec1 : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":2},{"id":"9e576056f6156fcf706156688ad576e0fb4e044e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 18:39:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aff39ffb7cac18eb8423dfc7783d1ad1c54cfb5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 18:40:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13863b6f77662e1809cee5da71f32b2ecfc7dce7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 20:23:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43d9da5a5824443d95b924d8ac9553fe\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ef4b8bd7a1244b8a5bb09921896a9ab : SUCCESS in 1h 42m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb476675dae4a0583bb442c3c9f3d55 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd09928293464429b71a341c4b0617f5 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdf905c6d9d84d4fab4f29e770c605e3 : SUCCESS in 2m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/270cd3cec6e643cba1e42e20afed33a3 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94830fd035af49079a17ccf880a29d60 : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5da688dbd3d642dc8e517344e84bbb24 : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":2},{"id":"da3a97d9271a1b0a7b8731647d10482c9f9cb6dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 20:23:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"01c8c7017e20e7691481ae122fc05b6ceb21df11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 20:25:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2dba57a147d451a9a48fb1d1734cac6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29504665532d41549ce703035e8cea2a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32217a238c89aa51f15aefe7da4cec183bcfa11e","revisions":{"4a30b32ec271f9d63b0aa62b3f9928ef80ef7083":{"kind":"REWORK","_number":1,"created":"2026-06-24 10:18:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/45/994645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/45/994645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/45/994645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/45/994645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/45/994645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/45/994645/1"}}},"commit":{"parents":[{"commit":"02007e69d85e90a9d6f9bd9d28841a45442849b1","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/02007e69d85e90a9d6f9bd9d28841a45442849b1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-06-24 10:15:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-06-24 10:15:29.000000000","tz":60},"subject":"zuul: Use openstack-python3-next-jobs template","message":"zuul: Use openstack-python3-next-jobs template\n\nTest the project against the latest and greatest Python versions.\n\nChange-Id: I9fe1dfe6bf139dc4e422d1b5ba4f7e8fe5f01d54\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/4a30b32ec271f9d63b0aa62b3f9928ef80ef7083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/4a30b32ec271f9d63b0aa62b3f9928ef80ef7083"}]},"branch":"refs/heads/master"},"32217a238c89aa51f15aefe7da4cec183bcfa11e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-24 12:48:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/45/994645/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.service","ref":"refs/changes/45/994645/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/45/994645/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/45/994645/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.service refs/changes/45/994645/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.service refs/changes/45/994645/2"}}},"commit":{"parents":[{"commit":"c360381ad9e5c1c2e3d9fb8f8f4b4ba6fe683703","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/c360381ad9e5c1c2e3d9fb8f8f4b4ba6fe683703"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-06-24 10:15:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-06-24 12:48:25.000000000","tz":60},"subject":"zuul: Use openstack-python3-next-jobs template","message":"zuul: Use openstack-python3-next-jobs template\n\nTest the project against the latest and greatest Python versions.\n\nChange-Id: I9fe1dfe6bf139dc4e422d1b5ba4f7e8fe5f01d54\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/32217a238c89aa51f15aefe7da4cec183bcfa11e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.service/commit/32217a238c89aa51f15aefe7da4cec183bcfa11e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
