)]}'
{"id":"openstack%2Foslo.tools~791438","triplet_id":"openstack%2Foslo.tools~master~I9496ef8ac2349e9bc5a3a426fb0fed4b25b95090","project":"openstack/oslo.tools","branch":"master","hashtags":[],"change_id":"I9496ef8ac2349e9bc5a3a426fb0fed4b25b95090","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-05-14 09:38:37.000000000","updated":"2021-05-14 11:23:14.000000000","submitted":"2021-05-14 11:23:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791438","meta_rev_id":"5b596dfddbba8033f23d14535d6803a24f683926","_number":791438,"virtual_id_number":791438,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 11:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-05-14 11:17:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-05-14 11:17:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 09:55:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 11:17:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"fd3f3b27f552b155f17f957c43bd03e4b9ec70cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-05-14 09:38:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44e616584bcd190d2d167722c0b0362b04f0d14e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 09:55:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3091a4f639140fdaf6e75ea63ce383d : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3386b84118c048d08f28b9f75c5717b0 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a56f1390c2db4f28b7fd5869cd441e23 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3c90d8275144c69ae9e1c3deff98c9a : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a270573402024cf69f54213d1ec3bb5f : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6181d44cf563bf4398eae796b4c6e896a51f7be7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-14 11:17:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"55d8faf5e9e3082642ee05734eebe6ff22857cc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 11:17:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"545984a6f910f20ba6cc2da8175e0b818b3b9342","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 11:23:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56955f9fd4494c368e3450d69db503f5 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c37579944744b8951fc2dc0add1ea9 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcee8163e14c4f43ad5c3d8044bffcbf : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0706b3210a48a4a20cb1857c35217a : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5b596dfddbba8033f23d14535d6803a24f683926","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 11:23:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dc78f0da135e166d2645b43d8794b174655b4d8","revisions":{"8dc78f0da135e166d2645b43d8794b174655b4d8":{"kind":"REWORK","_number":1,"created":"2021-05-14 09:38:37.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/38/791438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.tools","ref":"refs/changes/38/791438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/38/791438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/38/791438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/38/791438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.tools refs/changes/38/791438/1"}}},"commit":{"parents":[{"commit":"f9ed48e9205ca20bc925926d781ebd59e810d6f9","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/f9ed48e9205ca20bc925926d781ebd59e810d6f9"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-05-14 09:37:58.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-05-14 09:38:33.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I9496ef8ac2349e9bc5a3a426fb0fed4b25b95090\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/8dc78f0da135e166d2645b43d8794b174655b4d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/8dc78f0da135e166d2645b43d8794b174655b4d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
