)]}'
{"id":"openstack%2Foslo.tools~840610","triplet_id":"openstack%2Foslo.tools~master~I7997e1dc076ba4b9a385b96bfc97b7bf1c8aedd0","project":"openstack/oslo.tools","branch":"master","topic":"drop-py36","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-05-23 09:31:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7997e1dc076ba4b9a385b96bfc97b7bf1c8aedd0","subject":"Drop python3.6/3.7 support in testing runtime","status":"MERGED","created":"2022-05-05 09:37:44.000000000","updated":"2022-05-23 09:31:57.000000000","submitted":"2022-05-23 09:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840610-drop-py36","meta_rev_id":"0452bce1a8f97aa39619626eae7bb5eb3041d7bb","_number":840610,"virtual_id_number":840610,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-23 09:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-05-23 09:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-05-23 09:22:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 09:45:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 17:07:12.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"5900380a501dbb77f9efebc151c64deecef6590e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 09:37:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62e10443a658b100907303888f87d78271f59077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 09:45:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c3f6462ccd044d286c0137244a2cef7 : SUCCESS in 2m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7f3827ef4ef41fd9db2f22f9b42304c : FAILURE in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59225381198d4c649e9a5959ffc5b459 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a4f2167ed046e19f4e94dd16971b1c : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"06e20e8fbc6486ea4e99ad5e43cd2df11ae5d8f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 13:49:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1371718798b97114a6f172ff9496602c5745ec09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 13:54:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f88dafce2df47689371b74cda9e5c9b : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe6bc94506094c9d8d6aabcaf112bc06 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbe8db10d2874f5492008b737d08ba3c : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f20253c74e234a1c92b5ebce46bdaf27 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782bee59558f47d28e78357108982252 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"fd7972b585d5634b949797e358902e00202bbec9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-05 17:07:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39ce2393c88269eb9d592ca463ecf73fc7987500","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-09 13:11:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f6e2c5fdb113e15ac1a5cfaff99afb2a7183b77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 13:16:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118dda7612014f7dadc6f5ea694d09d9 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1deca8fb728f4172ba9852a74a22ae07 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edfdb642080445109d3c5f7b4fa6c137 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6738db460f7b46d19c5369b692ce2c1b : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f00c13ac4144dfaeed1fc3bdd1bf0d : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"969b69f4e0fcdfe322016e8019fd487c4332f3f0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-23 09:22:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d92a765d925724ae2ac590e9b1ada50f6d18ff7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 09:23:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"87c06cbd550dab597fe4efa9ab0bb1710c412215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 09:31:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d5ee48790746428b4c22ea2359760f : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60354b5cb4ec45ce8306d118bffde851 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1a2e88773ae46248e3b6af9b8ed47fb : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313ac7fc719b47868f2619c4c51af7d1 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"0452bce1a8f97aa39619626eae7bb5eb3041d7bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 09:31:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5a046ea4c9b069c54e60ed5e927d286159ae73a3","revisions":{"16b962d96d330a0cd31494886f83a3c2ec4cd241":{"kind":"REWORK","_number":1,"created":"2022-05-05 09:37:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/840610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.tools","ref":"refs/changes/10/840610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/1"}}},"commit":{"parents":[{"commit":"39e580a1e2bd5cd8c243c2a8681414fdbf04ca38","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/39e580a1e2bd5cd8c243c2a8681414fdbf04ca38"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 09:37:29.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 09:37:29.000000000","tz":120},"subject":"Drop python3.6 support in testing runtime","message":"Drop python3.6 support in testing runtime\n\nIn Zed cycle testing runtime, we are targetting to drop the\npython 3.6 support, project started adding python 3.8 as minimum,\n\nexample nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nChange-Id: I7997e1dc076ba4b9a385b96bfc97b7bf1c8aedd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/16b962d96d330a0cd31494886f83a3c2ec4cd241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/16b962d96d330a0cd31494886f83a3c2ec4cd241"}]},"branch":"refs/heads/master"},"e34662fc95560eef2e5fc4a9e44d6edafecdcb2d":{"kind":"REWORK","_number":2,"created":"2022-05-05 13:49:17.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/840610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.tools","ref":"refs/changes/10/840610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/2"}}},"commit":{"parents":[{"commit":"39e580a1e2bd5cd8c243c2a8681414fdbf04ca38","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/39e580a1e2bd5cd8c243c2a8681414fdbf04ca38"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 09:37:29.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 13:48:28.000000000","tz":120},"subject":"Drop python3.6/3.7 support in testing runtime","message":"Drop python3.6/3.7 support in testing runtime\n\nIn Zed cycle testing runtime, we are targetting to drop the\npython 3.6/3.7 support, project started adding python 3.8 as minimum,\n\nexample nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nAlso moving zuul\u0027s job to Zed.\n\nChange-Id: I7997e1dc076ba4b9a385b96bfc97b7bf1c8aedd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/e34662fc95560eef2e5fc4a9e44d6edafecdcb2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/e34662fc95560eef2e5fc4a9e44d6edafecdcb2d"}]},"branch":"refs/heads/master"},"5a046ea4c9b069c54e60ed5e927d286159ae73a3":{"kind":"REWORK","_number":3,"created":"2022-05-09 13:11:41.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/840610/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.tools","ref":"refs/changes/10/840610/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.tools refs/changes/10/840610/3"}}},"commit":{"parents":[{"commit":"39e580a1e2bd5cd8c243c2a8681414fdbf04ca38","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/39e580a1e2bd5cd8c243c2a8681414fdbf04ca38"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 09:37:29.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-09 13:11:20.000000000","tz":120},"subject":"Drop python3.6/3.7 support in testing runtime","message":"Drop python3.6/3.7 support in testing runtime\n\nIn Zed cycle testing runtime, we are targetting to drop the\npython 3.6/3.7 support, project started adding python 3.8 as minimum,\n\nexample nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nAlso moving zuul\u0027s job to `openstack-python3-jobs`.\n\nChange-Id: I7997e1dc076ba4b9a385b96bfc97b7bf1c8aedd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/5a046ea4c9b069c54e60ed5e927d286159ae73a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/5a046ea4c9b069c54e60ed5e927d286159ae73a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
