)]}'
{"id":"openstack%2Foslo.tools~882064","triplet_id":"openstack%2Foslo.tools~master~Iad5f31553c19f243be487fdcf8cadd9812839c51","project":"openstack/oslo.tools","branch":"master","topic":"oslo-drop-py38","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-05-05 12:06:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-05-05 12:17:53.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-05-05 12:17:53.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-05-05 12:17:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad5f31553c19f243be487fdcf8cadd9812839c51","subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","status":"MERGED","created":"2023-05-03 20:00:28.000000000","updated":"2023-05-05 12:17:53.000000000","submitted":"2023-05-05 12:17:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":879313,"submission_id":"882064-oslo-drop-py38","meta_rev_id":"d247e87e27d36945151eceea7f5857dd96bd259e","_number":882064,"virtual_id_number":882064,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-05 12:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-05-05 12:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-05-05 12:06:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-03 20:00:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-05-03 20:00:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-05-03 20:00:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-03 20:00:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3642a2c5b88815168b548013d87b5f0575b14eb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-03 20:00:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a71c0f3462d1fc6b1bea4e5490a040a2af05ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 20:08:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/707191d13d92459a8a17c6322c2db4f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d826ed32cba454b8757c151b2131082 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a868e41494cd4420bc805315826cb3f1 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d98eaa7a45f4cee9d6b09713acafb96 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2d94fc508d48c1a410afead03d187f : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f0b365e115087975120351c99f1823230964b552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-04 20:21:29.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"64b4507d491f4393aa9683f83d04a019c8324204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 20:34:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d4007a59ab847be8ae16d302870afc7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32b3296eb855472eb92cfd8eb323dc23 : SUCCESS in 2m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be641957248b412b99e81ae464316bda : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5060398781f4af1a57dae39c1c49521 : SUCCESS in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9a1848e33e4501876a6fc0cd66f77c : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":2},{"id":"f8cefa0d5ebb25a19ad097bb1b5ec6a466fb8410","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-05 12:06:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dff2bdaea966b585cc43908894630164c972553a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 12:07:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c12857e848b7c5cad73f1e3e9f0f9fb137339cd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 12:17:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7756a68f0f4c1991ac4ffdc38deaa1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c077e8ec55824854a38e019e9bc2b63c : SUCCESS in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab0cf70b4fba459a9c61220ac7f2fcce : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9aaa46e2ce54b77bf233b7a46f7014a : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495e0d8c8da54b7d95a4ca4eaf6cb6d3 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d247e87e27d36945151eceea7f5857dd96bd259e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 12:17:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a85b2e529994384057453a278c45aa970ed87ae1","revisions":{"6a9aacaf30149c25e2b13edeeeaf8af91f4d1358":{"kind":"REWORK","_number":1,"created":"2023-05-03 20:00:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/882064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.tools","ref":"refs/changes/64/882064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/64/882064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/64/882064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/64/882064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.tools refs/changes/64/882064/1"}}},"commit":{"parents":[{"commit":"9739bd0047d61166e52cc3f6bb6294aeac09477d","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/9739bd0047d61166e52cc3f6bb6294aeac09477d"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:00:28.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:00:28.000000000","tz":0},"subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","message":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n\nThis reverts commit 9739bd0047d61166e52cc3f6bb6294aeac09477d.\n\nReason for revert:\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nTC has been discussing about re-adding the python 3.8\ntesting in current master 2023.2 release testing.\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-04-25-18.00.log.html#l-191\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033469.html\n\nWhile governance changes are under review, TC agreed to add py3.8 testing\nso that we do not see more project/lib dropping python 3.8 and make them\nuninstalable on python 3.8\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-05-02-18.00.log.html#l-17\n- https://review.opendev.org/c/openstack/governance/+/882165\n\nAlso adding py3.8 testing back in job https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nChange-Id: Iad5f31553c19f243be487fdcf8cadd9812839c51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/6a9aacaf30149c25e2b13edeeeaf8af91f4d1358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/6a9aacaf30149c25e2b13edeeeaf8af91f4d1358"}]},"branch":"refs/heads/master"},"a85b2e529994384057453a278c45aa970ed87ae1":{"kind":"REWORK","_number":2,"created":"2023-05-04 20:21:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/882064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.tools","ref":"refs/changes/64/882064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/64/882064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/64/882064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/64/882064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.tools refs/changes/64/882064/2"}}},"commit":{"parents":[{"commit":"9739bd0047d61166e52cc3f6bb6294aeac09477d","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/9739bd0047d61166e52cc3f6bb6294aeac09477d"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:00:28.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-04 20:21:29.000000000","tz":0},"subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","message":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n\nThis reverts commit 9739bd0047d61166e52cc3f6bb6294aeac09477d.\n\nKeeping Python 3.10 in setup.cfg classifier. \n\nReason for revert:\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nTC has been discussing about re-adding the python 3.8\ntesting in current master 2023.2 release testing.\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-04-25-18.00.log.html#l-191\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033469.html\n\nWhile governance changes are under review, TC agreed to add py3.8 testing\nso that we do not see more project/lib dropping python 3.8 and make them\nuninstalable on python 3.8\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-05-02-18.00.log.html#l-17\n- https://review.opendev.org/c/openstack/governance/+/882165\n\nAlso adding py3.8 testing back in job https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nChange-Id: Iad5f31553c19f243be487fdcf8cadd9812839c51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/a85b2e529994384057453a278c45aa970ed87ae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/a85b2e529994384057453a278c45aa970ed87ae1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
