)]}'
{"id":"openstack%2Foslo.tools~898606","triplet_id":"openstack%2Foslo.tools~master~If112e3f4389fab8e5954ddf85a00a0615233740c","project":"openstack/oslo.tools","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-18 08:06:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If112e3f4389fab8e5954ddf85a00a0615233740c","subject":"Fix python shebang","status":"MERGED","created":"2023-10-17 15:51:39.000000000","updated":"2023-10-18 08:06:32.000000000","submitted":"2023-10-18 08:06:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898606","meta_rev_id":"bbaa17a81a5a9ee3ba153cdab962874afe081045","_number":898606,"virtual_id_number":898606,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-18 08:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2023-10-18 07:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2023-10-18 07:49:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-17 16:02:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-18 07:49:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"dbdd52284d0d30ee3212a6545fe782ac0a15c76d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-17 15:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"990e773320509fadaa84a9c248ec3223f5a659b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 16:02:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2783900e710c47d8a6ad0298bf484eeb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad39da9ab8624f0fb930f6c404ca46fc : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e81167ef64c44b8b9927ba00024c7301 : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59d3ec2e165c4c8ab307baef2b1982ab : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50ce540808e543a69f4ca2acb2a71821 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc780ded29744ae6bd5ae63c6ab88c52 : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20da88e4df5f46f89784116b51eb8d0b : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"24de1a81f6a1fc16a516e5e62b386d43d9611400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-17 16:02:54.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5defbedac8c533388188aa585a584cc9691dcb04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 16:18:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bd7c183dfe24315b39eeb773d25e2b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d75ae009b384c16a0fa9a3d1f6a525a : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7562186a46d04858ae1547d4b34a7a00 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57461fa5bc3a42abb2e16c3929ecf111 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfbd4c5136f044e2a5f308a056b48970 : SUCCESS in 2m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bbc851266bc4658938bd1f5e944138c : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8d3919c6f8433c8f005b20c28fe18c : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"13d8bedf0485ae4f5dd4ca79e3184ba6a689988e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-10-18 07:49:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1550947b91179dfbb82e6902b4b5beecf18cc8d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:50:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c2234f830d718d2d3bb21c2c0e2d6d7ae785384","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 08:06:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f3f7b609da14985992e5fa50e1b9b05\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3decb999fb4c4148848207da6d8ee620 : SUCCESS in 2m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8201647557664946a2986c8582e65ade : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53910c5e562044648b515453a9019977 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57856c44a9544a3a99c3473000151215 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48cad78aef7344f9bbbb27a6211c0a42 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"bbaa17a81a5a9ee3ba153cdab962874afe081045","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 08:06:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd5a039df30c8fe48c286fe2c1c4f9ebec54d98b","revisions":{"e0bfd5c126b04a6be0932360fb96ceb07d249d6c":{"kind":"REWORK","_number":1,"created":"2023-10-17 15:51:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/898606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.tools","ref":"refs/changes/06/898606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/06/898606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/06/898606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/06/898606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.tools refs/changes/06/898606/1"}}},"commit":{"parents":[{"commit":"a85b2e529994384057453a278c45aa970ed87ae1","subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/a85b2e529994384057453a278c45aa970ed87ae1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-17 15:51:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-17 15:51:37.000000000","tz":540},"subject":"Fix python shebang","message":"Fix python shebang\n\nThe current shebang requires python 2 instead of python 3.\n\nChange-Id: If112e3f4389fab8e5954ddf85a00a0615233740c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/e0bfd5c126b04a6be0932360fb96ceb07d249d6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/e0bfd5c126b04a6be0932360fb96ceb07d249d6c"}]},"branch":"refs/heads/master"},"cd5a039df30c8fe48c286fe2c1c4f9ebec54d98b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-17 16:02:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/898606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.tools","ref":"refs/changes/06/898606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/06/898606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/06/898606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.tools refs/changes/06/898606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.tools refs/changes/06/898606/2"}}},"commit":{"parents":[{"commit":"a85b2e529994384057453a278c45aa970ed87ae1","subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/a85b2e529994384057453a278c45aa970ed87ae1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-17 15:51:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-17 16:02:54.000000000","tz":0},"subject":"Fix python shebang","message":"Fix python shebang\n\nThe current shebang requires /usr/bin/python which is not available in\nUbuntu Jammy by default.\n\nChange-Id: If112e3f4389fab8e5954ddf85a00a0615233740c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/cd5a039df30c8fe48c286fe2c1c4f9ebec54d98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.tools/commit/cd5a039df30c8fe48c286fe2c1c4f9ebec54d98b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
