)]}'
{"id":"openstack%2Foslo.upgradecheck~652461","triplet_id":"openstack%2Foslo.upgradecheck~master~I67e81197ed63764f68b95c6d05c6d2c074929362","project":"openstack/oslo.upgradecheck","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"I67e81197ed63764f68b95c6d05c6d2c074929362","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-15 01:27:18.000000000","updated":"2019-05-03 16:43:15.000000000","submitted":"2019-05-03 16:43:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652461-1556901795477-0b0d46a4","meta_rev_id":"d5d34bec395ada13d7896f3a4811d6c2e5e7cbfa","_number":652461,"virtual_id_number":652461,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-03 16:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-05-03 16:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-04-23 14:32:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-05-03 16:33:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 14:32:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-03 16:33:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-03 16:43:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd6e75651d36e7f4d93df3a37ab9036598df017e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 01:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ecf2f1b1f3bf2f385143ab6bd892aef99542193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 01:34:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/61/652461/1/check/test-release-openstack/25d0637/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/61/652461/1/check/openstack-tox-docs/d4696c6/html/ : SUCCESS in 6m 31s\n- openstack-tox-pep8 http://logs.openstack.org/61/652461/1/check/openstack-tox-pep8/d07a68c/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/61/652461/1/check/openstack-tox-py27/f20994c/ : SUCCESS in 5m 00s\n- openstack-tox-py37 http://logs.openstack.org/61/652461/1/check/openstack-tox-py37/67e6092/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/61/652461/1/check/openstack-tox-py36/0aeda6a/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/652461/1/check/openstack-tox-lower-constraints/066081d/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2fa0a442941dc3b5f14ba18ad1037288ee9d679e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-23 14:32:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4cfe080dedd49a01fdbccc9de0fa54873a9217d1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:33:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"064f6a9627b9f5ff8064b6f007d5352797886715","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:33:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d064682b240005a1f91f847db9b6e2e88f068ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:43:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1eac6fc6833b7dcf03ae64c54587bd5b03e158b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:43:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/61/652461/1/gate/test-release-openstack/605cd82/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/61/652461/1/gate/openstack-tox-docs/1bb3d3f/html/ : SUCCESS in 6m 35s\n- openstack-tox-pep8 http://logs.openstack.org/61/652461/1/gate/openstack-tox-pep8/66fca9e/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/61/652461/1/gate/openstack-tox-py27/446add3/ : SUCCESS in 4m 00s\n- openstack-tox-py37 http://logs.openstack.org/61/652461/1/gate/openstack-tox-py37/ca749b2/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/61/652461/1/gate/openstack-tox-py36/cec1c55/ : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/652461/1/gate/openstack-tox-lower-constraints/6b9130a/ : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41f47aeba6f25f98b23763bba050aa60d652a928","revisions":{"41f47aeba6f25f98b23763bba050aa60d652a928":{"kind":"REWORK","_number":1,"created":"2019-04-15 01:27:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/61/652461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/61/652461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/61/652461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/61/652461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/61/652461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/61/652461/1"}}},"commit":{"parents":[{"commit":"b1a3426b32fe222067ea036b0b7d6fb7191edff3","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/b1a3426b32fe222067ea036b0b7d6fb7191edff3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:26:59.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:27:16.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I67e81197ed63764f68b95c6d05c6d2c074929362\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/41f47aeba6f25f98b23763bba050aa60d652a928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/41f47aeba6f25f98b23763bba050aa60d652a928"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
